)]}'
{"id":"openstack%2Freleases~682073","triplet_id":"openstack%2Freleases~master~Idc590a9940eeb1ecb367b08c2b623133aaf5f750","project":"openstack/releases","branch":"master","topic":"kuryr_train_deliverable","hashtags":[],"change_id":"Idc590a9940eeb1ecb367b08c2b623133aaf5f750","subject":"Release Kuryr-Kubernetes 1.1.0 and cut train","status":"MERGED","created":"2019-09-13 14:52:53.000000000","updated":"2019-09-16 07:59:53.000000000","submitted":"2019-09-15 19:01:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682073-1568574106757-9b7968ff","meta_rev_id":"502ee32f4c926dfcdb1dccb9ba0dd776e5b588c2","_number":682073,"virtual_id_number":682073,"owner":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-09-13 15:18:02.000000000","_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-09-15 19:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":2,"date":"2019-09-13 19:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":1,"date":"2019-09-14 11:51:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2019-09-14 09:24:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":1,"date":"2019-09-14 00:02:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":1,"date":"2019-09-15 18:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 14:53:16.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2019-09-13 14:53:19.000000000","updated_by":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"reviewer":{"_account_id":14352,"name":"Antoni Segura Puimedon","email":"celebdor@gmail.com","username":"toni"},"state":"REVIEWER"},{"updated":"2019-09-13 14:53:25.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-09-13 14:53:30.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2019-09-13 15:04:52.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-13 15:09:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-14 00:02:07.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-09-14 09:24:45.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-09-14 11:51:28.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-09-15 18:51:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-15 19:01:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b556da00bbd74128235eeb0b724a2247faa2cb9f","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-09-13 14:52:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7a94c4deb7b254917e4a5a620bd0b1e47b0d7db","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-13 15:09:20.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM needs Michal\u0027s review","accounts_in_message":[],"_revision_number":1},{"id":"0cb0f439e1f66988029d4956bffbbe6bfc89c53c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 15:11:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74a6fe60face851d991836d3632e70d943b8ec53","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-09-13 15:17:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"160913bcb0f774a12631dda1dbc5a010bf5d1cc4","author":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"date":"2019-09-13 15:18:02.000000000","message":"Patch Set 2:\n\nwell, not that it\u0027s that much work from my side to add it (fixing a typo anyway) ;)","accounts_in_message":[],"_revision_number":2},{"id":"ba3d2f269f4070338b6d96d0d3432e96cc33de9f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 15:26:41.000000000","message":"Patch Set 2:\n\nThanks!\n\nJust need the ack from Michal yet.","accounts_in_message":[],"_revision_number":2},{"id":"5c1f51e54342725e73749b684dcf534400aa84a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:35:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54451d678884391a5ce0cdbac5432e0 : SUCCESS in 9m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a0c74b5d8b540e7b05fdf193ce9765a : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64646be4ba6b41199934557a45f7eeed : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":2},{"id":"cd734d5b1072eab878788e176b3007e83f92dbc7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-13 19:09:23.000000000","message":"Patch Set 2: Code-Review+2\n\nLast commit has not changed since the one here. Doesn\u0027t look like there are any other patches about to merge, so since we are past the deadline we should probably just merge this. Will wait a little bit longer to see if we can get an explicit all clear.","accounts_in_message":[],"_revision_number":2},{"id":"3ebc162a7b80db5d71e885c561b9107596c0e5dc","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-09-14 00:02:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"26359270314b0d761f5bcee1e15f633b4844dbc3","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-09-14 09:24:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f06d1348f1c302188eef88233ab82975fdcf74bd","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-09-14 11:51:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5138f2c2eee741178db844aba1a4835068f5929e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-15 18:51:49.000000000","message":"Patch Set 2: Workflow+1\n\nLooks safe to assume Michal is OK with this, so since we\u0027re three days past the deadline I\u0027m going to approve.","accounts_in_message":[],"_revision_number":2},{"id":"dbd83dbaf73605f2b4ea8311ac69db7a222cf385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 18:52:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c27dac1b7ab7e73b56f4af81a21236e8ac5a0991","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:01:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8d5574aad12ca0ed8ec47b173dd198edf895aff0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 19:01:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbe95b38ab14ba791bc7c3eeb3b9165 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96b23394228a4160a30098a0ba2c8be5 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1b2f933f44d5eeb6e8c7e3d9b3fc657fbc794b7f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-09-16 07:59:53.000000000","message":"Patch Set 2:\n\nYes, sorry, I was out for a few days. This is totally okay, thanks for taking care of it!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dc6ec2c5a77cd2721f80b7a4853c2e3af0a43713","revisions":{"96c89853dff0895429349bd810d2340bcd158ed9":{"kind":"REWORK","_number":1,"created":"2019-09-13 14:52:53.000000000","uploader":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"ref":"refs/changes/73/682073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/682073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/682073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/682073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/682073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/682073/1"}}},"commit":{"parents":[{"commit":"8cee4966fa11c3a54801e347137df0193416fa24","subject":"Merge \"Auto releasing python-saharaclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8cee4966fa11c3a54801e347137df0193416fa24"}]}],"author":{"name":"Daniel Mellado","email":"dmellado@redhat.com","date":"2019-09-13 14:52:17.000000000","tz":120},"committer":{"name":"Daniel Mellado","email":"dmellado@redhat.com","date":"2019-09-13 14:52:51.000000000","tz":120},"subject":"Release Kuryr-Kubernetes 1.1.0 and cut train","message":"Release Kuryr-Kubernetes 1.1.0 and cut train\n\nChange-Id: Idc590a9940eeb1ecb367b08c2b623133aaf5f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96c89853dff0895429349bd810d2340bcd158ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96c89853dff0895429349bd810d2340bcd158ed9"}]},"branch":"refs/heads/master"},"dc6ec2c5a77cd2721f80b7a4853c2e3af0a43713":{"kind":"REWORK","_number":2,"created":"2019-09-13 15:17:33.000000000","uploader":{"_account_id":14885,"name":"Daniel Mellado","email":"dmellado@redhat.com","username":"daniel-mellado"},"ref":"refs/changes/73/682073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/682073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/682073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/682073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/682073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/682073/2"}}},"commit":{"parents":[{"commit":"8cee4966fa11c3a54801e347137df0193416fa24","subject":"Merge \"Auto releasing python-saharaclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8cee4966fa11c3a54801e347137df0193416fa24"}]}],"author":{"name":"Daniel Mellado","email":"dmellado@redhat.com","date":"2019-09-13 14:52:17.000000000","tz":120},"committer":{"name":"Daniel Mellado","email":"dmellado@redhat.com","date":"2019-09-13 15:15:47.000000000","tz":120},"subject":"Release Kuryr-Kubernetes 1.1.0 and cut train","message":"Release Kuryr-Kubernetes 1.1.0 and cut train\n\nChange-Id: Idc590a9940eeb1ecb367b08c2b623133aaf5f750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc6ec2c5a77cd2721f80b7a4853c2e3af0a43713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc6ec2c5a77cd2721f80b7a4853c2e3af0a43713"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
