)]}'
{"id":"openstack%2Freleases~682347","triplet_id":"openstack%2Freleases~master~I7fc8c89b3bb684ffdb434587125ab48fc68dd21d","project":"openstack/releases","branch":"master","topic":"os-win-stein","hashtags":[],"change_id":"I7fc8c89b3bb684ffdb434587125ab48fc68dd21d","subject":"Release os-win 4.2.1 (stein)","status":"MERGED","created":"2019-09-16 11:50:11.000000000","updated":"2019-09-19 10:02:36.000000000","submitted":"2019-09-19 10:02:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682347-1568887356534-5c82e828","meta_rev_id":"4cd9da690043cfbde703b8cc42654dae54bda3a5","_number":682347,"virtual_id_number":682347,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":2,"date":"2019-09-19 10:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"date":"2019-09-16 12:36:49.000000000","_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-16 16:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-19 09:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-16 13:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 13:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-19 09:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 13:03:20.000000000","updated_by":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"reviewer":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"state":"REVIEWER"},{"updated":"2019-09-16 16:47:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-18 13:29:47.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-19 09:47:18.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-19 10:02:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1b9a273028e33f46bf550e97e2cdb4c4124ce5d","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-16 11:50:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bd62a3f9b6956e14ba83150ab44e0791614b7db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 12:02:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb648b6417e4b5ca1a75f9ab4f85c58 : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/482c3e3d826349e79712183e9841066a : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e8d8223564cf44abbcbb447cb982e22c : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5fbf9c12cbe690fbd7b22fe5fbc6367a28798a7e","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-09-16 12:33:14.000000000","message":"Patch Set 1: Code-Review-1\n\nCommit hash is identical to the previous release.","accounts_in_message":[],"_revision_number":1},{"id":"53bc7444a82be6be52d05c1297a39f767b748f68","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-16 12:36:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99f1c743f113a4efd14473621ee8ebe5a95803db","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-16 12:36:49.000000000","message":"Patch Set 2:\n\nWrong paste, thanks for noticing!","accounts_in_message":[],"_revision_number":2},{"id":"306f7285d95357571d5958af3df4b94148aeb58a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 12:51:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e5b2a993494c759130130abe9964d2 : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f02fbd6d5904105900ec5cfe9a02b95 : SUCCESS in 5m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bafec6d836ca4247aba940aedbc45e8e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b0ff5f9a8981ed1513a26a3ac4cb88e12e36da50","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2019-09-16 13:03:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4bde443de9d572e7534ed8d894733156b1455560","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-16 16:47:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49af5331351f2f9d492291bd40af234e3880f39c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-18 13:29:47.000000000","message":"Patch Set 2: Code-Review+2\n\nOk for me, as it\u0027s bugfixing.","accounts_in_message":[],"_revision_number":2},{"id":"169aef9f59900bd486c643e73a679450bf86928e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-19 09:47:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa649e517bb96cefcc2bd7a6e6dc7c09e78fe1b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 09:47:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6d07cc3e2a0766250cc64d81497e3b7feef2334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 10:02:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"00d690dc1d17c95e0d51ccd97797045db4156c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 10:02:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea031bef7206481faf2ca9567978ee81 : SUCCESS in 8m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37e836de78b54558864ff5b76a859843 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9ef291006c532861c2cb33729bfd60dfebfd3f3","revisions":{"a0331c4d56cbe6749ac85a6d81ce01a4902346cd":{"kind":"REWORK","_number":1,"created":"2019-09-16 11:50:11.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/47/682347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/682347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/682347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/682347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/682347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/682347/1"}}},"commit":{"parents":[{"commit":"0858e754bc786b57c73ac0af52df5087b7375b20","subject":"Merge \"Release puppet-nova 12.5.0 (queens)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0858e754bc786b57c73ac0af52df5087b7375b20"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-16 11:44:06.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-16 11:44:06.000000000","tz":180},"subject":"Release os-win 4.2.1 (stein)","message":"Release os-win 4.2.1 (stein)\n\nChange-Id: I7fc8c89b3bb684ffdb434587125ab48fc68dd21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0331c4d56cbe6749ac85a6d81ce01a4902346cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0331c4d56cbe6749ac85a6d81ce01a4902346cd"}]},"branch":"refs/heads/master"},"d9ef291006c532861c2cb33729bfd60dfebfd3f3":{"kind":"REWORK","_number":2,"created":"2019-09-16 12:36:25.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/47/682347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/682347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/682347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/682347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/682347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/682347/2"}}},"commit":{"parents":[{"commit":"0858e754bc786b57c73ac0af52df5087b7375b20","subject":"Merge \"Release puppet-nova 12.5.0 (queens)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0858e754bc786b57c73ac0af52df5087b7375b20"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-16 11:44:06.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2019-09-16 12:36:06.000000000","tz":180},"subject":"Release os-win 4.2.1 (stein)","message":"Release os-win 4.2.1 (stein)\n\nChange-Id: I7fc8c89b3bb684ffdb434587125ab48fc68dd21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9ef291006c532861c2cb33729bfd60dfebfd3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9ef291006c532861c2cb33729bfd60dfebfd3f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
