)]}'
{"id":"openstack%2Freleases~682509","triplet_id":"openstack%2Freleases~master~I1c98e4ef45aadc391921a0420ecbcaebeecb987f","project":"openstack/releases","branch":"master","topic":"nova-train-cycle-highlights","hashtags":[],"change_id":"I1c98e4ef45aadc391921a0420ecbcaebeecb987f","subject":"Add highlight for VPMEM in Train release","status":"MERGED","created":"2019-09-16 22:55:33.000000000","updated":"2019-09-18 14:44:15.000000000","submitted":"2019-09-18 14:44:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682509-1568817855052-3f030d00","meta_rev_id":"13f7f50f4800126146fe97516a3bc9db6d03c694","_number":682509,"virtual_id_number":682509,"owner":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-09-18 14:36:24.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-09-18 14:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-17 21:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-17 13:51:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-09-18 00:00:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 14:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-18 14:33:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 13:51:42.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-17 21:10:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-18 00:00:04.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-09-18 14:33:32.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-18 14:36:24.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-18 14:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ebdbdc7953ddc3db93febce84132ee5a7809eb3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-09-16 22:55:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b05801eaeaedb69410535e15010e1aa2ab6ccd46","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-09-16 22:57:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e844cafb263c42d08f3f0117e6d829d4aaf435ce","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-16 22:59:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5572e1a53dc45f8fe1b287bb0f37f662fdecf06b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-16 23:06:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c405ee475e08db4b8ac55cc16a3522b361279212","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:14:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9ae839c6fd4d3b8a3d225bde8f60e7 : SUCCESS in 9m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a946baa21d994fdb8151e372e92d4ab6 : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/714214d4350e44ada9217f8d7eec672c : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2d3081b68121b9d6f094c96158a474f54d6aa5ff","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-09-16 23:46:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b28c88a4558420e9009a55d806cb21b9933637a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:01:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7fc9fb445444b2984b28ec71ea13c0 : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12a10fb8610940aa89d4d366fcf3c23d : SUCCESS in 4m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5f180ff678a4231a76938319d6385a4 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5183ccacb0651eb27f3f8fc8bff41e6a31462241","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-17 05:42:27.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"31c3c78509e2c4b3e51b7be592ab6d5c445094c4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-17 12:25:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"428569866832f77a591665d4a200c5aee7247866","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-17 12:26:29.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks Alex. I reworded a bit to make the sentence structure flow and match that of the other bullets. This lgtm.","accounts_in_message":[],"_revision_number":4},{"id":"716780719c4babc4f42b047a1c2860026943a886","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 12:41:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14840b57952d4307b3337bc5f6e70891 : SUCCESS in 9m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83fc7a3bdf284cf1803fa50361f14aeb : SUCCESS in 4m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/703f2f17e8564012b13c365f95565105 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"a5036da7f26109613c6541b07c8bdc164b412f7f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 13:17:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2668335a0c3560a535114712b85be2ff203ac771","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-17 13:24:17.000000000","message":"Patch Set 4: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ef099f97a7e272d50f245515fae36233ae71c581","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-17 13:35:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"50bd8b051486a524b99c6545ae7f7452a0452920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:47:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a5efccbfd24811a2b5f0c3b964f555 : SUCCESS in 8m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/447f288b4d144f9e8177e2e8a63dbbcf : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e2ad9bac75ea410ab931ce6a226b617a : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":5},{"id":"96cd4172c9349d71e83acaa830e65f4cf2b65f3d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-17 13:51:42.000000000","message":"Patch Set 5: Code-Review+1\n\nbetter https://openstack.fortnebula.com:13808/v1/AUTH_e8fd161dc34c421a979a9e6421f823e9/zuul_opendev_logs_69a/682509/5/check/openstack-tox-docs/69a5efc/docs/train/highlights.html#nova-compute-service","accounts_in_message":[],"_revision_number":5},{"id":"c2787a496ea5e98506262775ee66bdb77661b602","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-17 21:10:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e832384c919da71a662ecb990a1e84d6abe30768","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-09-18 00:00:04.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ea993a5504d707273ee9c4b58eed08807aa94d85","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-18 14:33:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nAs Eric is fine with the content, I am fine with merging this.","accounts_in_message":[],"_revision_number":5},{"id":"f0f313bb02051443bd47c25ec8a75791ad8dc9b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:33:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1e1a4e7b5483f3ed8d393ecc2385aa23110fa941","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-18 14:36:24.000000000","message":"Patch Set 5:\n\nNote that the vpmem stuff hasn\u0027t fully merged in nova yet either but it sounds like we\u0027re committed at this point.","accounts_in_message":[],"_revision_number":5},{"id":"14c606c22b2e250cd6a4ccf8f4cfc86abc9a6850","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:44:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aefbdf407fa84e1c8d3e3e6f32ca631c : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/491eb1e840fc43ed9b02db91b5fcba90 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":5},{"id":"90479d2a6524c48430361a6ff9966c45e8277230","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 14:44:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c3d30898fc1e1b70243a0497e59a1059d3fcf091","revisions":{"8e3dad7f4499941e3c3c6aa4897c93c684649c87":{"kind":"REWORK","_number":1,"created":"2019-09-16 22:55:33.000000000","uploader":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"ref":"refs/changes/09/682509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/682509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/682509/1"}}},"commit":{"parents":[{"commit":"78b8bf44cb03cbe945d155fb097396a22e53994c","subject":"Merge \"Release kuryr 1.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78b8bf44cb03cbe945d155fb097396a22e53994c"}]}],"author":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:10:29.000000000","tz":480},"committer":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:10:29.000000000","tz":480},"subject":"Add hightlight for VPMEM in Train release","message":"Add hightlight for VPMEM in Train release\n\nChange-Id: I1c98e4ef45aadc391921a0420ecbcaebeecb987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e3dad7f4499941e3c3c6aa4897c93c684649c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e3dad7f4499941e3c3c6aa4897c93c684649c87"}]},"branch":"refs/heads/master"},"508d5b330a99b689c1dec31ccd8173f9b646b383":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-16 22:57:23.000000000","uploader":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"ref":"refs/changes/09/682509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/682509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/682509/2"}}},"commit":{"parents":[{"commit":"78b8bf44cb03cbe945d155fb097396a22e53994c","subject":"Merge \"Release kuryr 1.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78b8bf44cb03cbe945d155fb097396a22e53994c"}]}],"author":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:10:29.000000000","tz":480},"committer":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:11:37.000000000","tz":480},"subject":"Add highlight for VPMEM in Train release","message":"Add highlight for VPMEM in Train release\n\nChange-Id: I1c98e4ef45aadc391921a0420ecbcaebeecb987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/508d5b330a99b689c1dec31ccd8173f9b646b383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/508d5b330a99b689c1dec31ccd8173f9b646b383"}]},"branch":"refs/heads/master"},"25d4e5cfaf221cf046597169da05c0134930ffb5":{"kind":"REWORK","_number":3,"created":"2019-09-16 23:46:01.000000000","uploader":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"ref":"refs/changes/09/682509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/682509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/682509/3"}}},"commit":{"parents":[{"commit":"78b8bf44cb03cbe945d155fb097396a22e53994c","subject":"Merge \"Release kuryr 1.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78b8bf44cb03cbe945d155fb097396a22e53994c"}]}],"author":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:10:29.000000000","tz":480},"committer":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-17 00:01:09.000000000","tz":480},"subject":"Add highlight for VPMEM in Train release","message":"Add highlight for VPMEM in Train release\n\nChange-Id: I1c98e4ef45aadc391921a0420ecbcaebeecb987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25d4e5cfaf221cf046597169da05c0134930ffb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25d4e5cfaf221cf046597169da05c0134930ffb5"}]},"branch":"refs/heads/master"},"f2ba4b3ca735d6a402c40bff7447215cf0ff4d4e":{"kind":"REWORK","_number":4,"created":"2019-09-17 12:25:43.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/09/682509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/682509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/682509/4"}}},"commit":{"parents":[{"commit":"78b8bf44cb03cbe945d155fb097396a22e53994c","subject":"Merge \"Release kuryr 1.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78b8bf44cb03cbe945d155fb097396a22e53994c"}]}],"author":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:10:29.000000000","tz":480},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-17 12:24:56.000000000","tz":-300},"subject":"Add highlight for VPMEM in Train release","message":"Add highlight for VPMEM in Train release\n\nChange-Id: I1c98e4ef45aadc391921a0420ecbcaebeecb987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2ba4b3ca735d6a402c40bff7447215cf0ff4d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2ba4b3ca735d6a402c40bff7447215cf0ff4d4e"}]},"branch":"refs/heads/master"},"c3d30898fc1e1b70243a0497e59a1059d3fcf091":{"kind":"REWORK","_number":5,"created":"2019-09-17 13:35:38.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/09/682509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/682509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/682509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/682509/5"}}},"commit":{"parents":[{"commit":"78b8bf44cb03cbe945d155fb097396a22e53994c","subject":"Merge \"Release kuryr 1.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78b8bf44cb03cbe945d155fb097396a22e53994c"}]}],"author":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2019-09-16 23:10:29.000000000","tz":480},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-17 13:35:36.000000000","tz":-300},"subject":"Add highlight for VPMEM in Train release","message":"Add highlight for VPMEM in Train release\n\nChange-Id: I1c98e4ef45aadc391921a0420ecbcaebeecb987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3d30898fc1e1b70243a0497e59a1059d3fcf091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3d30898fc1e1b70243a0497e59a1059d3fcf091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
