)]}'
{"id":"openstack%2Freleases~682656","triplet_id":"openstack%2Freleases~master~I517880906019ea52f9b3731319abcad55f92915d","project":"openstack/releases","branch":"master","topic":"sqlalchemy-migrate-0.13.0","hashtags":[],"change_id":"I517880906019ea52f9b3731319abcad55f92915d","subject":"ussuri: release sqlalchemy-migrate 0.13.0","status":"MERGED","created":"2019-09-17 14:10:16.000000000","updated":"2019-10-22 21:04:13.000000000","submitted":"2019-10-22 21:04:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682656-1571778253411-2a96c748","meta_rev_id":"cf4c0a43a96bf783fe68c2e959a3fc1c713b30a9","_number":682656,"virtual_id_number":682656,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-10-21 17:23:34.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 21:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-21 16:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-22 18:34:21.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-22 20:52:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 18:34:21.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-10-22 20:52:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-22 21:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b049b7b50d0db315d5d45a79a9587c3dfa24e006","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 14:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0576efd0fab0a1325b357ba4de33b9c3cace1896","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 14:10:45.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for Ussuri to open.","accounts_in_message":[],"_revision_number":1},{"id":"9770ab5470c08467f08057fd313fc4b7477f48ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 14:17:37.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"fb1ef22b44c880b7779a86464b126adfccfcf88e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 18:29:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9cc46b39f3afbc7d2a8dd9e96e4b5c7e18e3c150","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 18:29:19.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c9558d80e03c2bb6f80a4b01974e2f7084021f81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:42:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4ef5dfa9ee4853a0471954819bb575 : SUCCESS in 11m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e505f72a5ec4bee99f03acb2d42b318 : FAILURE in 6m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1ab27fae3c11408cbd21272418fa2337 : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"bdc883c99adb997d46f2096c47d4977353e3c41b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-21 14:32:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd453ded9abfe983cdee4ba33da8f5fccac15990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:58:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac79952761c487a873291ff7094a56f : SUCCESS in 9m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/755f101804464763bc711c43cdd5ec8e : FAILURE in 6m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dfd7016c0fd84c3e9a12759a00dd4391 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"0860eab42351697ab9014dc2eb867553f4484c9a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-21 15:39:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0531865afcba864aade322d4c054316e77c713d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 15:52:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42db295c28d247a3a3a77df3e64e27ec : SUCCESS in 8m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e70962e8ed474ce18b24a12bc850bd4b : SUCCESS in 5m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d87a2e1e62054b3ea65fde61870a600e : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":4},{"id":"a7edc6aaefe9764d255d1d2b1ed93bea023dd8f3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-21 16:16:33.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)\n\nShould be fine. We can always fix any issues if we find a problem later.","accounts_in_message":[],"_revision_number":4},{"id":"81eb8824a7cd54a7b4808e449c11831141fc096c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-21 17:23:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc7371a9321a5d65be9e6b071f853de736ae001a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-10-22 18:34:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nLooks good to me","accounts_in_message":[],"_revision_number":4},{"id":"8a93d2c40428b22bd1bb79386c4987ebdb21c6e4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 20:52:55.000000000","message":"Patch Set 4: Workflow+1\n\n/me crosses fingers","accounts_in_message":[],"_revision_number":4},{"id":"88c9c4492c807bbe0a360033be99cec28bb7c315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 20:53:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0b20a9ce56563bba2f75d0dda63c0da58eb280dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 21:04:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fefa49d26190074649d41f545185bfd974323b3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 21:04:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a7dce9dc744cf5ad70bead947b3338 : SUCCESS in 8m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a945370129064f3aacb6f359eb3bb959 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ebda55aafb983755c166c8aaf39debf04d085a69","revisions":{"e9f4ed2ee079bf2dd194ff19fbdbff62e6f123d3":{"kind":"REWORK","_number":1,"created":"2019-09-17 14:10:16.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/56/682656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/682656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/682656/1"}}},"commit":{"parents":[{"commit":"9b5b0d67e3fe30e55339558d89d51faba4acd558","subject":"Merge \"Add deliverable file for Kayobe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b5b0d67e3fe30e55339558d89d51faba4acd558"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 14:08:01.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 14:09:55.000000000","tz":-240},"subject":"WIP: ussuri: release sqlalchemy-migrate 0.13.0","message":"WIP: ussuri: release sqlalchemy-migrate 0.13.0\n\nThis release most just contains cleanups and officially\nadds support for python 3.5 and 3.6. It uses mysqlclient\nrather than the old MySQL-Python.\n\nDepends-On: https://review.opendev.org/666857/\n\nChange-Id: I517880906019ea52f9b3731319abcad55f92915d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9f4ed2ee079bf2dd194ff19fbdbff62e6f123d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9f4ed2ee079bf2dd194ff19fbdbff62e6f123d3"}]},"branch":"refs/heads/master"},"8070ba74ca09e1240fc0a84f4ba9e2c1a63aa025":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-17 18:29:06.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/56/682656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/682656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/682656/2"}}},"commit":{"parents":[{"commit":"9b5b0d67e3fe30e55339558d89d51faba4acd558","subject":"Merge \"Add deliverable file for Kayobe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b5b0d67e3fe30e55339558d89d51faba4acd558"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 14:08:01.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 18:28:52.000000000","tz":-240},"subject":"WIP: ussuri: release sqlalchemy-migrate 0.13.0","message":"WIP: ussuri: release sqlalchemy-migrate 0.13.0\n\nThis release most just contains cleanups and officially\nadds support for python 3.5 and 3.6. It uses mysqlclient\nrather than the old MySQL-Python.\n\nDepends-On: https://review.opendev.org/682518/\n\nChange-Id: I517880906019ea52f9b3731319abcad55f92915d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8070ba74ca09e1240fc0a84f4ba9e2c1a63aa025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8070ba74ca09e1240fc0a84f4ba9e2c1a63aa025"}]},"branch":"refs/heads/master"},"dd440fd57f572e4b24b4e5dba45d23a9094730af":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2019-10-21 14:32:23.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/56/682656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/682656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/682656/3"}}},"commit":{"parents":[{"commit":"e4adf7b08e298c716bb46abdf6439a62996fc302","subject":"Merge \"TripleO RC tags + stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4adf7b08e298c716bb46abdf6439a62996fc302"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 14:08:01.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-21 14:32:05.000000000","tz":-240},"subject":"ussuri: release sqlalchemy-migrate 0.13.0","message":"ussuri: release sqlalchemy-migrate 0.13.0\n\nThis release most just contains cleanups and officially\nadds support for python 3.5 and 3.6. It uses mysqlclient\nrather than the old MySQL-Python.\n\nChange-Id: I517880906019ea52f9b3731319abcad55f92915d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd440fd57f572e4b24b4e5dba45d23a9094730af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd440fd57f572e4b24b4e5dba45d23a9094730af"}]},"branch":"refs/heads/master"},"ebda55aafb983755c166c8aaf39debf04d085a69":{"kind":"REWORK","_number":4,"created":"2019-10-21 15:39:08.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/56/682656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/682656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/682656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/682656/4"}}},"commit":{"parents":[{"commit":"e4adf7b08e298c716bb46abdf6439a62996fc302","subject":"Merge \"TripleO RC tags + stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4adf7b08e298c716bb46abdf6439a62996fc302"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 14:08:01.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-21 15:38:43.000000000","tz":-240},"subject":"ussuri: release sqlalchemy-migrate 0.13.0","message":"ussuri: release sqlalchemy-migrate 0.13.0\n\nThis release most just contains cleanups and officially\nadds support for python 3.5 and 3.6. It uses mysqlclient\nrather than the old MySQL-Python.\n\nNote this is the first release since the repo moved from\nthe openstack to x namespace in opendev.\n\nChange-Id: I517880906019ea52f9b3731319abcad55f92915d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ebda55aafb983755c166c8aaf39debf04d085a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ebda55aafb983755c166c8aaf39debf04d085a69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
