)]}'
{"id":"openstack%2Freleases~683398","triplet_id":"openstack%2Freleases~master~Icfa9a3d6a5b0678132d274b557354df824328ec8","project":"openstack/releases","branch":"master","topic":"magnum-8.1.0","hashtags":[],"change_id":"Icfa9a3d6a5b0678132d274b557354df824328ec8","subject":"Release magnum 8.1.0","status":"MERGED","created":"2019-09-20 14:15:22.000000000","updated":"2019-10-01 12:52:14.000000000","submitted":"2019-10-01 12:52:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683398-1569934334413-02882d20","meta_rev_id":"ff39047031b122247438aa9f30a37a9a75f1e94d","_number":683398,"virtual_id_number":683398,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 12:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-27 10:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-27 10:54:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-27 10:54:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":2,"date":"2019-10-01 12:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2019-09-27 11:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":1,"date":"2019-10-01 12:34:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 14:16:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-26 12:20:10.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2019-09-26 12:56:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-27 03:29:01.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2019-09-27 10:38:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-27 11:23:34.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-10-01 12:34:17.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-01 12:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bdad6d2d908e71289740941b055a2f32c728b10","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-20 14:15:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4810ff005967a8a2a198f494f6b680419125f1e","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-20 14:15:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"430ae12cb074850406d4050a8139c4414cfc7d9c","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-20 14:16:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"302818a6c073e453c0b6568eb0c661e0088c3cd1","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-20 14:22:04.000000000","message":"Patch Set 2: Workflow-1\n\nWaiting for the back ported commits to merge.","accounts_in_message":[],"_revision_number":2},{"id":"abd24fd3207460f1249c186295fff7ebdae511a9","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-20 14:23:12.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/#/c/683387/","accounts_in_message":[],"_revision_number":2},{"id":"17e356776e06453bf75bb37770ef8ddb69d16d76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:32:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34991cfcd8504f4f8a3dae1af171b2f2 : SUCCESS in 8m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b27afa5101a4b3cb9e4f3e478e58fe2 : FAILURE in 4m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bc5de998eaf84fc9ad1e85c27421259c : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"440b5abb3a3ce4c46be8388784619f4dbb5bc79e","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-26 10:51:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e1a754fe79752efa392527e211fdf345fafa4f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 11:08:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01da4bfc742744aa9f4a280a94081956 : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afedae8b0e0344b489be10cd4b90f80d : SUCCESS in 5m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1c5641c22c97402eb2fa72f719a6d0d2 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"cfc3a561e52a426ff125da1ef18b029bbac8853f","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2019-09-26 12:20:10.000000000","message":"Patch Set 3: Code-Review+1\n\nflwang should approve","accounts_in_message":[],"_revision_number":3},{"id":"3bee9c2b18fc5179c9f04fd25015ccd636c73795","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 12:56:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3ae084dca0c5e20fe6d22592b010133241cad6a4","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2019-09-27 03:29:01.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"a7f5b0685716cd04e0f090d053fa391d4dd0ca3b","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-27 10:36:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a6373d4738b776c4289490b63e1e713d2a62ec5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-27 10:38:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2b73c5cc2ea716bb29e882ca1c016c0660a4ea92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:50:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27aa0391871f438c8dac414a684f6924 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0b9d81f95234b24a58270ff827d3451 : FAILURE in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/57717414e44343eb9079bbc14eae0662 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":4},{"id":"876f8124d5ef6c7ac16839889c948043842b5f60","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-27 10:54:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ec88e95ca792eb2102e05e38790f5f6712d00d76","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-09-27 11:08:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1269db3c2ec2256d55a7e894708001c001c9b6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 11:10:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc6cab9ae8d4f869566bdc191ca3e73 : SUCCESS in 8m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9b14c94853546749353ec73c20688a9 : SUCCESS in 4m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/84015bfd9dd4435e9934672a166f90a0 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":5},{"id":"c065ab8a1e6edb5e173700567adfbb15204a9e1c","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-09-27 11:23:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"de7f3b9f45d1f17508570516490b0c7cf4149b9d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-01 12:34:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f14de6afc20aafec9047716ff7b79cb56d1b1e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:34:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"83a3a898bf4f9d7dfeee6ed87813a05be98a8941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:52:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3420db684294fd9acf20a0daea810b0 : SUCCESS in 9m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73e31f54c9064971b81649394104206a : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":5},{"id":"e20abe32b8785555ee9a9cf833c3456a0fd242ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:52:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1658094fe6aed376632358bbccdf1dcb65ef59ae","revisions":{"b5283459ddca84d86e6947ad84f1da354715b7c7":{"kind":"REWORK","_number":1,"created":"2019-09-20 14:15:22.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/98/683398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/683398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/683398/1"}}},"commit":{"parents":[{"commit":"36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7","subject":"Neutron highlights for Train.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-20 14:15:06.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-20 14:15:20.000000000","tz":0},"subject":"Release magnum 8.1.0","message":"Release magnum 8.1.0\n\nChange-Id: Icfa9a3d6a5b0678132d274b557354df824328ec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5283459ddca84d86e6947ad84f1da354715b7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5283459ddca84d86e6947ad84f1da354715b7c7"}]},"branch":"refs/heads/master"},"08ccb15f5e067539f566d359bd0843738cf50d7b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-20 14:15:58.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/98/683398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/683398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/683398/2"}}},"commit":{"parents":[{"commit":"36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7","subject":"Neutron highlights for Train.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-20 14:15:06.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-20 14:15:48.000000000","tz":0},"subject":"Release magnum 8.1.0","message":"Release magnum 8.1.0\n\nChange-Id: Icfa9a3d6a5b0678132d274b557354df824328ec8\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08ccb15f5e067539f566d359bd0843738cf50d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/08ccb15f5e067539f566d359bd0843738cf50d7b"}]},"branch":"refs/heads/master"},"e77379b4837ee5c91f5393bb2288c60a9a8aa606":{"kind":"REWORK","_number":3,"created":"2019-09-26 10:51:12.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/98/683398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/683398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/683398/3"}}},"commit":{"parents":[{"commit":"36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7","subject":"Neutron highlights for Train.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-20 14:15:06.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-26 10:51:01.000000000","tz":0},"subject":"Release magnum 8.1.0","message":"Release magnum 8.1.0\n\nChange-Id: Icfa9a3d6a5b0678132d274b557354df824328ec8\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e77379b4837ee5c91f5393bb2288c60a9a8aa606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e77379b4837ee5c91f5393bb2288c60a9a8aa606"}]},"branch":"refs/heads/master"},"633591a5381aa382bb3b1957cd484b4f67dfb5d4":{"kind":"REWORK","_number":4,"created":"2019-09-27 10:36:28.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/98/683398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/683398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/683398/4"}}},"commit":{"parents":[{"commit":"36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7","subject":"Neutron highlights for Train.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-20 14:15:06.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-27 10:36:19.000000000","tz":0},"subject":"Release magnum 8.1.0","message":"Release magnum 8.1.0\n\nChange-Id: Icfa9a3d6a5b0678132d274b557354df824328ec8\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/633591a5381aa382bb3b1957cd484b4f67dfb5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/633591a5381aa382bb3b1957cd484b4f67dfb5d4"}]},"branch":"refs/heads/master"},"1658094fe6aed376632358bbccdf1dcb65ef59ae":{"kind":"REWORK","_number":5,"created":"2019-09-27 10:54:28.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/98/683398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/683398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/683398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/683398/5"}}},"commit":{"parents":[{"commit":"36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7","subject":"Neutron highlights for Train.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/36ef64762d28d1a395ac3dc9fbfa67d376cbb4e7"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-27 10:54:17.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-09-27 10:54:17.000000000","tz":0},"subject":"Release magnum 8.1.0","message":"Release magnum 8.1.0\n\nChange-Id: Icfa9a3d6a5b0678132d274b557354df824328ec8\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1658094fe6aed376632358bbccdf1dcb65ef59ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1658094fe6aed376632358bbccdf1dcb65ef59ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
