)]}'
{"id":"openstack%2Freleases~683422","triplet_id":"openstack%2Freleases~master~I9ba636a8a3147466a1e661c074a050ff14a6d394","project":"openstack/releases","branch":"master","topic":"document_release-test_purpose","hashtags":[],"change_id":"I9ba636a8a3147466a1e661c074a050ff14a6d394","subject":"Clarify release-test purpose","status":"MERGED","created":"2019-09-20 15:15:07.000000000","updated":"2019-10-07 14:24:36.000000000","submitted":"2019-10-07 14:24:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683422-1570458276082-33a7a8e1","meta_rev_id":"b33b9e39002ec782389f438f84836536b6044211","_number":683422,"virtual_id_number":683422,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 14:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-07 14:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-23 08:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-07 14:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 08:34:48.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-07 14:11:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-07 14:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a1cf769993778991ae437ca57bb35f64b4f7eed","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-20 15:15:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44d542d142a0e90c35f6f7518e7e4f2f5f20f980","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-20 15:17:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ef8f656eff3c6f6a0499d55685a83b130743de61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:31:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8d8cb573544939989c23adf8e525f6 : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":2},{"id":"28dc5da0c06844ca59eddc01b7647e3a16a08101","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-23 08:34:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c93869628e47df473846ab94f3235bc25bb245b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-07 14:11:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWe have used release-test for other things, mostly whenever we needed to validate the release machinery for whatever reason. But I think this is a good reminder to include.","accounts_in_message":[],"_revision_number":2},{"id":"5bdc274cbcba9c167f304955bd3728b71e9889d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 14:11:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d81addaa06c78bac6c8a020d65398e3cdbdab49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 14:24:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1583e95fcb046bcbd9e582a05451eca : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9a2c87d364451fc30233d53c4a9bf3c0d3469b8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 14:24:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0e9fb43816a2f4091f77964785c94eb287dfce1","revisions":{"7295673acee9214e413b98023ce18cafe2a09e23":{"kind":"REWORK","_number":1,"created":"2019-09-20 15:15:07.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/22/683422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/683422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/683422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/683422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/683422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/683422/1"}}},"commit":{"parents":[{"commit":"eaf3c31aee8bc0c17d02bd9afcfe2503e865bfed","subject":"Merge \"Add ocata-eol tags for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eaf3c31aee8bc0c17d02bd9afcfe2503e865bfed"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-20 15:13:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-20 15:13:22.000000000","tz":120},"subject":"Clarify release-test purpose","message":"Clarify release-test purpose\n\nThis is not clear what release-test is for.\nI expected the repo to be used for trying to release, as a canary\ntest, but only got a confirmation through IRC (as this was not\nwritten in the process, until the section with the final release).\n\nThis should clarify things on _when_ release-test start to matter.\n\nChange-Id: I9ba636a8a3147466a1e661c074a050ff14a6d394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7295673acee9214e413b98023ce18cafe2a09e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7295673acee9214e413b98023ce18cafe2a09e23"}]},"branch":"refs/heads/master"},"a0e9fb43816a2f4091f77964785c94eb287dfce1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-20 15:17:29.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/22/683422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/683422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/683422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/683422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/683422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/683422/2"}}},"commit":{"parents":[{"commit":"eaf3c31aee8bc0c17d02bd9afcfe2503e865bfed","subject":"Merge \"Add ocata-eol tags for kolla and kolla-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eaf3c31aee8bc0c17d02bd9afcfe2503e865bfed"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-20 15:13:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-20 15:17:27.000000000","tz":0},"subject":"Clarify release-test purpose","message":"Clarify release-test purpose\n\nThis is not clear what release-test is for.\n\nI expected the repo to be used for final release, as a canary\ntest. I expected things had to done with it beforehand.\nBut I only got a confirmation through IRC (as this is not yet documented\nin the process), that it only serves during RC/final release time.\n\nThis should clarify it.\n\nChange-Id: I9ba636a8a3147466a1e661c074a050ff14a6d394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0e9fb43816a2f4091f77964785c94eb287dfce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0e9fb43816a2f4091f77964785c94eb287dfce1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
