)]}'
{"id":"openstack%2Freleases~683695","triplet_id":"openstack%2Freleases~master~I8d6a0c701331652b3049e7aeeba4ac9b57e11ffc","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I8d6a0c701331652b3049e7aeeba4ac9b57e11ffc","subject":"Release senlin-dashboard","status":"MERGED","created":"2019-09-20 18:56:48.000000000","updated":"2019-09-25 10:26:17.000000000","submitted":"2019-09-25 10:26:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683695-1569407177201-9cafdce8","meta_rev_id":"e710e40de77bd58603e8e25a9579121e10b903d0","_number":683695,"virtual_id_number":683695,"owner":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-09-24 07:30:51.000000000","_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 10:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"all":[{"value":1,"date":"2019-09-25 02:08:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":2,"date":"2019-09-25 08:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-25 10:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-25 10:09:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 02:08:00.000000000","updated_by":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"reviewer":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"state":"REVIEWER"},{"updated":"2019-09-25 08:38:15.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-25 10:09:01.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-25 10:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"541042fc9f43af3f8ec38c8922855f5bd1a52a75","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-20 18:56:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b31d07354291e581e16b65eb7e1817145b4e3d6c","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-20 19:00:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66715fbf2934e19c4236248e43833591ee735529","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 19:16:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82803f9db24a4bbeba8c63c5e0392b37 : SUCCESS in 9m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e90283fe9ac462e825b5b73b3681382 : FAILURE in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fcc3b7efc2e64f2db13f6e180abe9993 : FAILURE in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"472f6f281cc91a7a8b25cb65d97853de6395e308","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-20 19:51:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d8e93593a2c055d16a1d8a8f3081c54f38417611","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 20:17:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c56adb559f54904a19be2e1c50b8a57 : SUCCESS in 9m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9718fa4a1b094e28855fa3d3c5f37ff3 : FAILURE in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/58cfd9c508da45bcb9c370625c38339e : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4391cf5fd4386d6357ff0e581675ee6ea58e71af","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-22 04:45:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"11b1e6bbc623480f9876b15859fc647d858e8ba3","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-22 04:48:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe1aa2f69bae743fa24604cd59590d385ea934cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 05:24:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b573a2d05c4948a484660f8c286d51 : SUCCESS in 8m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a18537862788470cb8170398f8a754a5 : FAILURE in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/be58a6ff663745d0be4708f328b697da : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"5b91c352ee24ce61adc83fd3aae8a9786e76dac3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-23 08:37:10.000000000","message":"Patch Set 3:\n\nCommit 58515f732a6474e0035ede3c408caf3faa3670dc does not seem to exist (yet) in senlin-dashboard ?","accounts_in_message":[],"_revision_number":3},{"id":"4decf5d3a9905f00c62640d4a337ee2e2a0fd4c8","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-24 02:20:31.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1f44a7a722ae35c1e7510bda548ab742bbb959fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 02:35:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6da04daf694780bd58b496266da1f7 : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/893cabf53e3740dcb7951b3708b5ff09 : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/364d75f20b8a410da9bf2af46123ae42 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"f4648e52919caf0c9a396433d25fdf52d1b15ec3","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2019-09-24 07:30:51.000000000","message":"Patch Set 4:\n\n\u003e Commit 58515f732a6474e0035ede3c408caf3faa3670dc does not seem to\n \u003e exist (yet) in senlin-dashboard ?\n\nYes, that patch is no need to be merged.","accounts_in_message":[],"_revision_number":4},{"id":"324a91ec1ca5473f72e2868693a1c5cbb8b916e1","author":{"_account_id":16352,"name":"Shu Muto","email":"shu.mutow@gmail.com","username":"shu.mutow"},"date":"2019-09-25 02:08:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cda93e807d3d7c8c305c1377e9bb407b597c56fa","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-25 08:38:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"137704de00004244d95d51783b4eead2176b36e3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-25 10:09:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ad2284a28c8ecabd1f6a60e2d7815efb3f673c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:09:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87bcfc494d65c2b48127af4dd511f7b68647c620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:26:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a237cb7ad5df40e2a98e6b3ef131ba37 : SUCCESS in 9m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50e78d742da34bed888bd7f24d12c962 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":4},{"id":"e710e40de77bd58603e8e25a9579121e10b903d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:26:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7cd8d8a10e5a02e2ba980f789b02e049adacf5ec","revisions":{"ffaf015b945efbdb5cee6125e0ced31830c608e9":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:56:48.000000000","uploader":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"ref":"refs/changes/95/683695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/683695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/683695/1"}}},"commit":{"parents":[{"commit":"5274a7b9737eb172103c0a0110c5f19a2dc3e5bb","subject":"Merge \"Add missing stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5274a7b9737eb172103c0a0110c5f19a2dc3e5bb"}]}],"author":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-20 18:55:07.000000000","tz":480},"committer":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-20 18:55:07.000000000","tz":480},"subject":"Release senlin-dashboard","message":"Release senlin-dashboard\n\nChange-Id: I8d6a0c701331652b3049e7aeeba4ac9b57e11ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ffaf015b945efbdb5cee6125e0ced31830c608e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ffaf015b945efbdb5cee6125e0ced31830c608e9"}]},"branch":"refs/heads/master"},"31f170103886781352e37d871d3278d0d074cd94":{"kind":"REWORK","_number":2,"created":"2019-09-20 19:00:14.000000000","uploader":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"ref":"refs/changes/95/683695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/683695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/683695/2"}}},"commit":{"parents":[{"commit":"5274a7b9737eb172103c0a0110c5f19a2dc3e5bb","subject":"Merge \"Add missing stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5274a7b9737eb172103c0a0110c5f19a2dc3e5bb"}]}],"author":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-20 18:55:07.000000000","tz":480},"committer":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-20 18:59:30.000000000","tz":480},"subject":"Release senlin-dashboard","message":"Release senlin-dashboard\n\nChange-Id: I8d6a0c701331652b3049e7aeeba4ac9b57e11ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31f170103886781352e37d871d3278d0d074cd94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31f170103886781352e37d871d3278d0d074cd94"}]},"branch":"refs/heads/master"},"b8f3a393bc04791d322756f51fb3c8a69c425732":{"kind":"REWORK","_number":3,"created":"2019-09-22 04:48:56.000000000","uploader":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"ref":"refs/changes/95/683695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/683695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/683695/3"}}},"commit":{"parents":[{"commit":"5274a7b9737eb172103c0a0110c5f19a2dc3e5bb","subject":"Merge \"Add missing stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5274a7b9737eb172103c0a0110c5f19a2dc3e5bb"}]}],"author":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-20 18:55:07.000000000","tz":480},"committer":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-22 04:47:31.000000000","tz":480},"subject":"Release senlin-dashboard","message":"Release senlin-dashboard\n\nChange-Id: I8d6a0c701331652b3049e7aeeba4ac9b57e11ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8f3a393bc04791d322756f51fb3c8a69c425732"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8f3a393bc04791d322756f51fb3c8a69c425732"}]},"branch":"refs/heads/master"},"7cd8d8a10e5a02e2ba980f789b02e049adacf5ec":{"kind":"REWORK","_number":4,"created":"2019-09-24 02:20:31.000000000","uploader":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"ref":"refs/changes/95/683695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/683695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/683695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/683695/4"}}},"commit":{"parents":[{"commit":"5274a7b9737eb172103c0a0110c5f19a2dc3e5bb","subject":"Merge \"Add missing stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5274a7b9737eb172103c0a0110c5f19a2dc3e5bb"}]}],"author":{"name":"jonnary","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-20 18:55:07.000000000","tz":480},"committer":{"name":"XueFeng Liu","email":"liu.xuefeng1@zte.com.cn","date":"2019-09-24 02:20:28.000000000","tz":0},"subject":"Release senlin-dashboard","message":"Release senlin-dashboard\n\nChange-Id: I8d6a0c701331652b3049e7aeeba4ac9b57e11ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7cd8d8a10e5a02e2ba980f789b02e049adacf5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7cd8d8a10e5a02e2ba980f789b02e049adacf5ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
