)]}'
{"id":"openstack%2Freleases~683860","triplet_id":"openstack%2Freleases~master~If2ac772601893b134b4462cbff1223dbed41bebf","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"If2ac772601893b134b4462cbff1223dbed41bebf","subject":"Release python-cloudkittyclient 3.1.0","status":"MERGED","created":"2019-09-23 08:04:37.000000000","updated":"2019-09-25 09:10:10.000000000","submitted":"2019-09-25 09:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683860-1569402610279-e6979eff","meta_rev_id":"40109617448d594579ee32eb158fb55df3ebf21c","_number":683860,"virtual_id_number":683860,"owner":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-09-24 07:35:30.000000000","_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 09:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-09-23 15:41:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-09-23 16:11:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":2,"date":"2019-09-25 08:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":1,"date":"2019-09-25 08:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 15:41:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-23 16:11:56.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-25 08:57:48.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-09-25 09:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc1ad46a6db1cdc3e022835e70a74dd9bc51a2c8","author":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"date":"2019-09-23 08:04:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"604d04312111d16aa3dae1854c2c2144528bb93e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:21:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258c24b63cae47d7912fc3d9f40b1a7c : SUCCESS in 10m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75339e27dc2f43eaaa4d61d7039464da : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fe367f5ff78e438e80ca67328def20c2 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c2ebe3b61fd717e23b25d4791649525d48a1742b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 09:01:44.000000000","message":"Patch Set 1: Code-Review-1\n\nThe process would be to request a FFE, afaik, even if you don\u0027t have some other client or lib requiring you. Could you follow the process please?","accounts_in_message":[],"_revision_number":1},{"id":"9a9f222c9d55ea9390abc2753542c5ecf076cdc7","author":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"date":"2019-09-23 13:05:01.000000000","message":"Patch Set 1:\n\nHi Jean-Philippe. TBH we did not expect the stable/train to be branched from 3.0.0, as it matches an intermediary release. We need this client release in order to match cloudkitty 11.0. \n\nI can request an FFE, but to whom ? The release team ?\n\nThanks in advance for advice.","accounts_in_message":[],"_revision_number":1},{"id":"5b339baef0b15c8d2c357d789798f7e7a467ec42","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-23 14:46:41.000000000","message":"Patch Set 1:\n\nSince we are under requirements freeze (since the deadline to release client libraries 10 days ago), if this release is necessary for the train release we will need approval from the requirements team. Could you open a thread on the openstack-discuss ML asking for [requirements] to grant an exception ?","accounts_in_message":[],"_revision_number":1},{"id":"51632b83a1bd8ae2edb2ca33fdd51261ff4318de","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-23 15:41:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"314aeaaefa8f9d7130c72c48db5837f5e545e891","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-23 16:11:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"868a32436ea7a5025591c825679a0908a58208ba","author":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"date":"2019-09-24 07:35:30.000000000","message":"Patch Set 1:\n\nHi Jean-Philippe, as discussed on IRC, we\u0027ve asked for an FFE on the ML: http://lists.openstack.org/pipermail/openstack-discuss/2019-September/009596.html","accounts_in_message":[],"_revision_number":1},{"id":"5fd46bb009239d5974debf364b3c86128766efb3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-25 08:57:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8736a2421b42f957fab52df449c64888242b9e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 08:58:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"236b263b550aae60ae0b04fb8597992514b02378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:10:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07439cbcabaa4067bbca26e690af80a2 : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/508a4b1543fc457c8f87aa576a5d4f6b : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"40109617448d594579ee32eb158fb55df3ebf21c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:10:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70051ec84e23f47d7511080bdda45d5081915b13","revisions":{"70051ec84e23f47d7511080bdda45d5081915b13":{"kind":"REWORK","_number":1,"created":"2019-09-23 08:04:37.000000000","uploader":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"ref":"refs/changes/60/683860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/683860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/683860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/683860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/683860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/683860/1"}}},"commit":{"parents":[{"commit":"5274a7b9737eb172103c0a0110c5f19a2dc3e5bb","subject":"Merge \"Add missing stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5274a7b9737eb172103c0a0110c5f19a2dc3e5bb"}]}],"author":{"name":"Luka Peschke","email":"luka.peschke@objectif-libre.com","date":"2019-09-23 08:00:54.000000000","tz":120},"committer":{"name":"Luka Peschke","email":"luka.peschke@objectif-libre.com","date":"2019-09-23 08:00:54.000000000","tz":120},"subject":"Release python-cloudkittyclient 3.1.0","message":"Release python-cloudkittyclient 3.1.0\n\nChange-Id: If2ac772601893b134b4462cbff1223dbed41bebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/70051ec84e23f47d7511080bdda45d5081915b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/70051ec84e23f47d7511080bdda45d5081915b13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
