)]}'
{"id":"openstack%2Freleases~684055","triplet_id":"openstack%2Freleases~master~I9b251f575ee80656a2f07cb1921945bb8a0fef78","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I9b251f575ee80656a2f07cb1921945bb8a0fef78","subject":"Auto release request for congress","status":"MERGED","created":"2019-09-23 18:08:59.000000000","updated":"2019-09-26 09:19:59.000000000","submitted":"2019-09-26 09:19:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684055-1569489599941-654589fa","meta_rev_id":"e1a3143cdf41e46aacd7ba9fe94b46f48eb6745e","_number":684055,"virtual_id_number":684055,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 09:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2019-09-25 22:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-09-26 09:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-25 21:22:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-09-26 09:05:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 21:22:39.000000000","updated_by":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"reviewer":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"state":"REVIEWER"},{"updated":"2019-09-25 22:41:37.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-09-26 09:05:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-26 09:19:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5fc05e182d72f2bf233caa8f7d136cf2163c93a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:08:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6b5b2b8cd5624c9fd8654198db8c212332ad186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:31:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770028ae8b8040aaa67e42973cea1f5a : SUCCESS in 9m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/663ed2adafc649a287aa6c3beebc75b5 : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/62a20fb354984cb7a6845bb5306613b5 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6b373d92a8f93f14c89d674094fbf27d46b0e4da","author":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"date":"2019-09-25 21:22:39.000000000","message":"Patch Set 1: Code-Review+1\n\nGood to go.","accounts_in_message":[],"_revision_number":1},{"id":"341ad89b3d936d3225d28b4663afbd97f74e9361","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-09-25 22:41:37.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks! Looks good.","accounts_in_message":[],"_revision_number":1},{"id":"b78e05ab3f341ac4d73606e30a3fffb4eaf3f8b9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-26 09:05:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee058aada6fc24c07e240890b551869b0f724276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:05:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e345c3aa8ddbad1e0f92bf075868fbcc9e81bafe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:19:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825e12f78383477f8c6ac74ae62e2889 : SUCCESS in 10m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/069a1b4187074ff29ad09e0922f33f46 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"790938d6bbccbc2d0401282a03cb0b57b8a600a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:19:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3217bfadc8fe76ffbcb99ec54b727efa085e245","revisions":{"a3217bfadc8fe76ffbcb99ec54b727efa085e245":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:08:59.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/55/684055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/684055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/684055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/684055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/684055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/684055/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:08:52.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:08:52.000000000","tz":120},"subject":"Auto release request for congress","message":"Auto release request for congress\n\nChange-Id: I9b251f575ee80656a2f07cb1921945bb8a0fef78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3217bfadc8fe76ffbcb99ec54b727efa085e245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3217bfadc8fe76ffbcb99ec54b727efa085e245"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
