)]}'
{"id":"openstack%2Freleases~684056","triplet_id":"openstack%2Freleases~master~I00f50aa097574a615c50195ac4e40207149a906d","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I00f50aa097574a615c50195ac4e40207149a906d","subject":"Auto release request for cyborg","status":"MERGED","created":"2019-09-23 18:09:21.000000000","updated":"2019-09-27 08:57:04.000000000","submitted":"2019-09-27 08:57:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684056-1569574624107-efb49f7a","meta_rev_id":"65656ddda4f333450e87d04433616c0cc04b8e58","_number":684056,"virtual_id_number":684056,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 08:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-27 08:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-27 04:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-09-27 08:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 18:09:21.000000000","updated_by":{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},"reviewer":{"_account_id":13629,"name":"Zhipeng","email":"huangzhipeng@huawei.com","username":"zhipengh"},"state":"REVIEWER"},{"updated":"2019-09-26 19:58:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-27 04:42:32.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-09-27 08:41:18.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-27 08:57:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"501f0479d5c5b2b8cf83e782c4bba520d2134b37","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:09:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a6df3cf2ae6abaff538b541d5b3567bd0028dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:29:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dbeb455bc024841bb9f231b9d13f43e : SUCCESS in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/819e80a35e904bbb8566568186362e7e : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c470a8eb62fa4f139fdcdaf1c96ebd2b : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"411bd592b7657b9c6d551551ae5ba6360ab32223","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-23 18:45:37.000000000","message":"Patch Set 1: Code-Review-1\n\nWe have few more changes as WIP. I beleieve the last day is Thursday midnight. is it midnight UTC?","accounts_in_message":[],"_revision_number":1},{"id":"f251d47ff85e8a76107c9645bfffce19fcd56be3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-24 08:29:50.000000000","message":"Patch Set 1:\n\nIt\u0027s end of day Thursday... meaning that we approve the patch on Friday UTC morning unless a specific exception is granted.","accounts_in_message":[],"_revision_number":1},{"id":"fecb3118f47c01536f92397aecea99b61e9291b5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-25 19:57:37.000000000","message":"Patch Set 1:\n\nWe have following issues affecting RC:\n* https://review.opendev.org/682565: affects interaction with other services\n* https://review.opendev.org/673957: affects security\n* https://review.opendev.org/683733: affects security\n* https://review.opendev.org/684193: affects backwards compatibility","accounts_in_message":[],"_revision_number":1},{"id":"3a39c9ebbabfd1b13b1d66c18d8e9dabfb591f0a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 12:25:20.000000000","message":"Patch Set 1:\n\nAny updates on the progress of those Sundar? Is this going to meet today\u0027s deadline, or should we expect to have to wait on this?","accounts_in_message":[],"_revision_number":1},{"id":"0a9ed2d0dd6edf82937ab5c21688965d6be77610","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-26 14:50:08.000000000","message":"Patch Set 1:\n\nSean, we made lots of progress:\nhttps://review.opendev.org/682565: affects interaction with other services --\u003e close, but not done yet\nhttps://review.opendev.org/673957: affects security --\u003e merged\nhttps://review.opendev.org/683733: affects security --\u003e merged\nhttps://review.opendev.org/684193: affects backwards compatibility --\u003e merged\n\nWe also made progress on:\nhttps://review.opendev.org/#/c/681005/ -- this fixes a deviation from the approved specs, so we are doing our best to get this in.\n\nUnfortunately, we haven\u0027t got many eyeballs on documentation in general:\nhttps://review.opendev.org/#/c/677715/ --\u003e Add driver support matrix\nI am trying to close on that. Nearly all of our developers are in APAC region, so I\u0027ll have to wait till evening Pacific time to get a handle on that.\n\nDoes it look like Cyborg is the only laggard? Do others need more time too?","accounts_in_message":[],"_revision_number":1},{"id":"fea539df7fe02066efe4a815f04093dfa590b956","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 15:31:04.000000000","message":"Patch Set 1:\n\nThere will probably be others. Just to help track what is going on, do you think this will be ready for today\u0027s deadline, or do you think this will take a few days yet?\n\nKeep in mind, additional RCs can be done before the final deadline. But we also don\u0027t want to create the RC and branch of we know things are coming soon.","accounts_in_message":[],"_revision_number":1},{"id":"bb96b57ffd9966e24b48892de9f435ed42ecd62a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-26 16:36:07.000000000","message":"Patch Set 1:\n\nIMO, the pending tasks will take a few days to a week, for us to do them properly.","accounts_in_message":[],"_revision_number":1},{"id":"540a70cabfc090cfca88c3e3621719d641207778","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 17:33:01.000000000","message":"Patch Set 1:\n\n\u003e IMO, the pending tasks will take a few days to a week, for us to do\n \u003e them properly.\n\nHmm, in that case, it may make sense to cut an RC now to let downstream folks pick things up and start testing, then plan on cutting an rc2 once the other fixes get merged. Assuming the current state is not completely broken.","accounts_in_message":[],"_revision_number":1},{"id":"59f0d9b6b27bbe1ea0daf4cbfaa0ecf6403975a9","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-26 17:59:21.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nAgreed. Please use the latest hash.","accounts_in_message":[],"_revision_number":1},{"id":"287fdfdb5bffdc22ca547dac4acfe51ea0c95f5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 18:19:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb34c414e6a9f8d75b660ff1fde2ea83993c3d97","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-26 19:28:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf584a262bcf116c61483fa2e1f579430ca57f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:39:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e284072b1784b098e48494fbb703b90 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19d54547acf94953a8fb5e5dc11ee3f0 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a332dd42b0f6452bb08b0c522e9d9bd4 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d5ad3bb6750dfa499500d08b3167abd2b25d87a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 19:58:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1eed320a161ca317013ed233ed1dd5aed1424a5","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-27 01:07:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0ba5b5280390ff4753597c4734fa56c8575a17e1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-27 04:21:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d72d564052ee8df174146ebf3576e5f4596cb97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:34:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2ed2eaf28942b299b63b11bf542399 : SUCCESS in 7m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10218454ec3b43fdb32d7ba710484f89 : SUCCESS in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8f11cb4ab03d47a7a288c73e500fb84d : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ebedf7ef74bd295ffeebf2f6bb5ef92857d39ab2","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-09-27 04:42:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2dbce332745863f9c5b5800cdd57b3e36c1c6785","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 08:41:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4bafc16ee26ab2e84628110f31fd4facab36fa2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:41:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9496aa8d321c3a78c84b6ad076bfa1190a986809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:57:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d353d1116f4114b32727dd8f9a072e : SUCCESS in 9m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45ff1df51aff4e369af2a02626c430cb : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"4e6fccbd37cfdb629feda9844a3659f0c5fbb4b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:57:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb49f43e664977909acf9a79055671aaedc3fd3a","revisions":{"db2f92f187ba1d4acdc8245045a20867a9b3f044":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:09:21.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/56/684056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/684056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/684056/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:09:12.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:09:12.000000000","tz":120},"subject":"Auto release request for cyborg","message":"Auto release request for cyborg\n\nChange-Id: I00f50aa097574a615c50195ac4e40207149a906d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db2f92f187ba1d4acdc8245045a20867a9b3f044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db2f92f187ba1d4acdc8245045a20867a9b3f044"}]},"branch":"refs/heads/master"},"fb9d959ed20061bb5eb5746570962a09d6108db5":{"kind":"REWORK","_number":2,"created":"2019-09-26 19:28:41.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/56/684056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/684056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/684056/2"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:09:12.000000000","tz":120},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-26 21:27:40.000000000","tz":-420},"subject":"Auto release request for cyborg","message":"Auto release request for cyborg\n\nChange-Id: I00f50aa097574a615c50195ac4e40207149a906d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb9d959ed20061bb5eb5746570962a09d6108db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb9d959ed20061bb5eb5746570962a09d6108db5"}]},"branch":"refs/heads/master"},"fb49f43e664977909acf9a79055671aaedc3fd3a":{"kind":"REWORK","_number":3,"created":"2019-09-27 04:21:03.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/56/684056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/684056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/684056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/684056/3"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:09:12.000000000","tz":120},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-09-27 06:19:50.000000000","tz":-420},"subject":"Auto release request for cyborg","message":"Auto release request for cyborg\n\nChange-Id: I00f50aa097574a615c50195ac4e40207149a906d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb49f43e664977909acf9a79055671aaedc3fd3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb49f43e664977909acf9a79055671aaedc3fd3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
