)]}'
{"id":"openstack%2Freleases~684064","triplet_id":"openstack%2Freleases~master~Ic6def256e1704ee80f7810f06d9032dddabfc65a","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"Ic6def256e1704ee80f7810f06d9032dddabfc65a","subject":"Auto release request for horizon","status":"MERGED","created":"2019-09-23 18:14:41.000000000","updated":"2019-09-27 08:52:29.000000000","submitted":"2019-09-27 08:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684064-1569574349415-d3498057","meta_rev_id":"920ccb97dc2538d2a72d8b4424009e04c7a1f7b9","_number":684064,"virtual_id_number":684064,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 08:52:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-27 08:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-09-27 08:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-27 08:39:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 18:32:34.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-09-27 08:39:39.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-27 08:39:45.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-27 08:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a2b236eaefb94c0622d88b95ca90e7d551b3908","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8420708bc3e82c4234c4f15beca6cebe5c2564a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:39:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df738e2fadb423b8f6e2de12ea35053 : SUCCESS in 8m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a85994c93b7c471191ad0fced0f17c66 : SUCCESS in 5m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c60634189aa04582bccdd3e111b10497 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"798553935e3e1301471e9b8490afcb477fae0739","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-24 18:38:27.000000000","message":"Patch Set 1:\n\nWe have some pending changes which are good for RC1.\nWe will discuss the status of them in the team meeting on Wednesday.","accounts_in_message":[],"_revision_number":1},{"id":"9898911edd4f0f8a5f31f982cc33ad52bba2826c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-26 18:32:34.000000000","message":"Patch Set 1:\n\nI\u0027m waiting for https://review.opendev.org/#/c/684792/ to be merged before propose RC1","accounts_in_message":[],"_revision_number":1},{"id":"943eed55d1d3982d7bca6e2d082a47a3b2a80d7f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-26 20:24:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3409805f4582f202dfd8fa8ef7b8132dee742512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 20:39:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfaee59c42d94679abbefb86ba473fe1 : SUCCESS in 10m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46cc90b8638e4ff58fe96a39d0fdb67b : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8c3a86cb55774d7abe8c49136909aed4 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4899cdf7cd0548eed3673367138248608c4feb43","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 08:37:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nConsidering Ivan\u0027s patch as a +1","accounts_in_message":[],"_revision_number":2},{"id":"86eb94329c67d18a014ed9a1ea630ebba882e399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:37:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2343216683f0609c4359ba42cc09aebf34be87d5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 08:37:51.000000000","message":"Patch Set 2: -Workflow\n\nHmm, needs amotoki\u0027s +1","accounts_in_message":[],"_revision_number":2},{"id":"4e05c73bd2ac44d10901e8908fc630600143075f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 08:39:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6044565ee79dbed72e685545b4739a24e8dad39c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-27 08:39:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3682e518cbd100d57fdbb0f1da4ceca822c51108","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 08:40:00.000000000","message":"Patch Set 2:\n\n(Ivan was PTL for train so it\u0027s ok)","accounts_in_message":[],"_revision_number":2},{"id":"8229868ab23d8c2adaee7552f616a5817d51106d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9cc04e0a1705b8e9f44244df51b88abe936e6d02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 08:52:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b394994aa654c4ba80ed5d4c74cf352 : SUCCESS in 9m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4a58f1de7364348bd40af85ba04e6ba : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c5d483e40d5f46a75e3b4f7451d57d19666d550","revisions":{"bbbde9385096d8fe12ee6cfb07f4693ebff26a14":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:14:41.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/64/684064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/684064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/684064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/684064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/684064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/684064/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:14:34.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:14:34.000000000","tz":120},"subject":"Auto release request for horizon","message":"Auto release request for horizon\n\nChange-Id: Ic6def256e1704ee80f7810f06d9032dddabfc65a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bbbde9385096d8fe12ee6cfb07f4693ebff26a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bbbde9385096d8fe12ee6cfb07f4693ebff26a14"}]},"branch":"refs/heads/master"},"6c5d483e40d5f46a75e3b4f7451d57d19666d550":{"kind":"REWORK","_number":2,"created":"2019-09-26 20:24:14.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/64/684064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/64/684064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/684064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/684064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/64/684064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/64/684064/2"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:14:34.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-09-26 20:23:56.000000000","tz":0},"subject":"Auto release request for horizon","message":"Auto release request for horizon\n\nChange-Id: Ic6def256e1704ee80f7810f06d9032dddabfc65a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c5d483e40d5f46a75e3b4f7451d57d19666d550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c5d483e40d5f46a75e3b4f7451d57d19666d550"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
