)]}'
{"id":"openstack%2Freleases~684079","triplet_id":"openstack%2Freleases~master~I7c62e6826806f8f424080bc043f3e4959bc63cd6","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I7c62e6826806f8f424080bc043f3e4959bc63cd6","subject":"Auto release request for networking-powervm","status":"MERGED","created":"2019-09-23 18:23:58.000000000","updated":"2019-09-26 09:25:06.000000000","submitted":"2019-09-26 09:25:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684079-1569489906965-b42f53c0","meta_rev_id":"1660aef8b7420f1f28e6fcafaa9149b54e27d578","_number":684079,"virtual_id_number":684079,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 09:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-26 09:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-26 04:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-09-26 05:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-26 09:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"date":"2019-09-24 10:39:33.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 10:39:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-09-26 04:00:39.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2019-09-26 05:12:31.000000000","updated_by":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"reviewer":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"state":"REVIEWER"},{"updated":"2019-09-26 09:06:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-26 09:25:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f0e81ce9b225fee845b7cce4bb1996251f26330","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52c0da689291063943925ab58c08ada797400d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:50:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e7adea09114ab89a8112253163f831 : SUCCESS in 9m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2943bea841f4f829592e9c0dfcb15bc : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/46adee9fdb8740859df89a9abe4d8ba7 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"268c2d55481955d5d8eb28fde53f951df3bc6996","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-24 10:39:33.000000000","message":"Patch Set 1:\n\nthis deliverable isn\u0027t managed by neutron team AFAIK so probably someone from PowerVMStackers team should review it","accounts_in_message":[],"_revision_number":1},{"id":"06441b6fdd78b512377dba1ae3b57d02bb39650b","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-09-26 04:00:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"21f2278fe2786c0eb8e6bff2e470a22e1d3e55fc","author":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"date":"2019-09-26 05:12:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"17be50019fac000e29300e6058de788d344b6fad","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-26 09:06:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d30794b6a9e1abd9cbcd4db45b82c17d7efdfe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:06:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4c1e21f69cd3722c071c712d417cea8a3048f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:25:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4579754e689a46d99b8417d70486b5f6 : SUCCESS in 10m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa66c56ebdaf4c06b0cacac84b34d865 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ca2226b7387b088a4f9b3d467ba33925b1c51ed7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:25:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a311694c0587a4daf3895f1c0a702162ca3f013","revisions":{"4a311694c0587a4daf3895f1c0a702162ca3f013":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:23:58.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/79/684079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/684079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/684079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/684079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/684079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/684079/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:23:51.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:23:51.000000000","tz":120},"subject":"Auto release request for networking-powervm","message":"Auto release request for networking-powervm\n\nChange-Id: I7c62e6826806f8f424080bc043f3e4959bc63cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a311694c0587a4daf3895f1c0a702162ca3f013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a311694c0587a4daf3895f1c0a702162ca3f013"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
