)]}'
{"id":"openstack%2Freleases~684084","triplet_id":"openstack%2Freleases~master~I71862cce395ab0a5e07ecb2993fc665259bc3934","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I71862cce395ab0a5e07ecb2993fc665259bc3934","subject":"Auto release request for neutron","status":"MERGED","created":"2019-09-23 18:29:09.000000000","updated":"2019-09-26 09:34:46.000000000","submitted":"2019-09-26 09:34:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684084-1569490486129-fe29816b","meta_rev_id":"2f0b5b9b8f492218503acec497a8c6601606327f","_number":684084,"virtual_id_number":684084,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 09:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":2,"date":"2019-09-25 22:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-09-26 09:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-26 07:18:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-09-25 21:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-09-26 09:07:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 21:39:57.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-25 22:39:01.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-09-26 07:18:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-09-26 09:07:25.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-26 09:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d71c7784484ea8a503ca57c7264c9bd2e677e8b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:29:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5491344d6852e988f1ddfabf3f212ba90add433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:57:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02df25678b14994acd67c589daad6f0 : SUCCESS in 9m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1c70abbd2a34f94b5f1cfe40ab5fc71 : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/589c74fa79d44c3f9dc213f75f0276be : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"98fb3ff4d2e014270250d0e9ccaace7fa82d4456","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-24 10:57:33.000000000","message":"Patch Set 1: Code-Review-1\n\nWe are still waiting for https://review.opendev.org/#/c/682418/","accounts_in_message":[],"_revision_number":1},{"id":"2acc94754e9248c03782c6acccbdcf96009e9da5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-25 20:29:05.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e We are still waiting for https://review.opendev.org/#/c/682418/\n\nThis patch is now finally merged, as well as one last missing releasenote: https://review.opendev.org/#/c/684361/\n\nSo we should be good to release rc-1 of neutron now.","accounts_in_message":[],"_revision_number":1},{"id":"d3a31ba48a85dff078c646b3586fb07a556d2ccd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-25 21:39:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e2dc14a00a57bff4c25950efa978b0f6c8c176bb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-25 21:39:57.000000000","message":"Patch Set 2: Code-Review+1\n\nI updated the commit hash for neutron RC1.","accounts_in_message":[],"_revision_number":2},{"id":"df6f427d936b2db093ed70a72d971cfda44b2c77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 21:58:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0677b1ed274b708ce4030fcd751a25 : SUCCESS in 10m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3adca5dc1b5048c5a7ce09c27191eedb : SUCCESS in 6m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7f5c613faadc47429b5e3a3272d1d661 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c80b7b26404ab9027d06c534b2784e90251774d0","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-09-25 22:39:01.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks! Looks good!","accounts_in_message":[],"_revision_number":2},{"id":"79c79ce92b586d4ab8dfe6f3fe0f204a38626295","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-26 07:18:35.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e I updated the commit hash for neutron RC1.\n\nThx. I forgot about it :/","accounts_in_message":[],"_revision_number":2},{"id":"9d82d6e8c2708769c490326775514398c5d55cf8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-26 09:07:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1b6bd79dea11344386eb1c770fd9ad40a1454ef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:07:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11a7c20184a307f6fdb8753f3294aef8c241f9c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:34:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef3434a218524ba5a5988ade574b9a02 : SUCCESS in 10m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1ed9a00675f49b4af8fec78d86d2ad3 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6bc76df2b8ea5a09de0f84a06ad433594731468f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:34:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5cabe356bb1e4d3a9b44436a88bb2240afda7bae","revisions":{"5534fcb4937708321a2b494102ce9f4cd5eb73f9":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:29:09.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/84/684084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/684084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/684084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/684084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/684084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/684084/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:29:01.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:29:01.000000000","tz":120},"subject":"Auto release request for neutron","message":"Auto release request for neutron\n\nChange-Id: I71862cce395ab0a5e07ecb2993fc665259bc3934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5534fcb4937708321a2b494102ce9f4cd5eb73f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5534fcb4937708321a2b494102ce9f4cd5eb73f9"}]},"branch":"refs/heads/master"},"5cabe356bb1e4d3a9b44436a88bb2240afda7bae":{"kind":"REWORK","_number":2,"created":"2019-09-25 21:39:37.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/84/684084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/84/684084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/684084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/684084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/84/684084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/84/684084/2"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:29:01.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-09-25 21:39:33.000000000","tz":0},"subject":"Auto release request for neutron","message":"Auto release request for neutron\n\nChange-Id: I71862cce395ab0a5e07ecb2993fc665259bc3934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cabe356bb1e4d3a9b44436a88bb2240afda7bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cabe356bb1e4d3a9b44436a88bb2240afda7bae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
