)]}'
{"id":"openstack%2Freleases~684088","triplet_id":"openstack%2Freleases~master~I7a0945bfd2a91eee4e105fc38235ae6784289bc0","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I7a0945bfd2a91eee4e105fc38235ae6784289bc0","subject":"Auto release request for placement","status":"MERGED","created":"2019-09-23 18:33:49.000000000","updated":"2019-09-26 09:34:44.000000000","submitted":"2019-09-26 09:34:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684088-1569490484798-d890bd4f","meta_rev_id":"a298dc093531391d9dc2049073d138d8d6d5ad83","_number":684088,"virtual_id_number":684088,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 09:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-25 19:44:40.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-26 09:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-09-25 22:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-09-25 19:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-09-25 19:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-09-26 05:40:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-26 09:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 19:44:40.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-09-25 19:48:33.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-25 19:53:51.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-25 22:41:54.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-09-26 05:40:54.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-09-26 09:06:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-26 09:34:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fc59e4ce2a10e0b24baa7459a6ecd8583575f5c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:33:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d732321004f5576210f86e72efc327953752ee21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 18:57:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798f3091115e408296a30a7a84eb6977 : SUCCESS in 9m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba51358af3564352ac8f8a7c1ba319d2 : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f64a90ff2be74e88b5660d4a3cf6ef8c : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4b2ae03d5d49799640ad6d4548e7197c0ad857a9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-23 21:14:40.000000000","message":"Patch Set 1:\n\n@JP, curious why Chris (Placement PTL) wasn\u0027t automatically added to the reviewer list for this.","accounts_in_message":[],"_revision_number":1},{"id":"267e143831f997b4d2579f4d7af7661c96c80003","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-23 22:11:14.000000000","message":"Patch Set 1:\n\n\u003e @JP, curious why Chris (Placement PTL) wasn\u0027t automatically added\n \u003e to the reviewer list for this.\n\nYou and Matt are listed as the release liaisons, I believe.\n\n(Will look at this properly tomorrow)","accounts_in_message":[],"_revision_number":1},{"id":"73774d027807233522cd113b7a9021b0b1a82333","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-23 23:32:50.000000000","message":"Patch Set 1:\n\n\u003e You and Matt are listed as the release liaisons, I believe.\n\nRight; I was told the PTL didn\u0027t need to be listed.","accounts_in_message":[],"_revision_number":1},{"id":"c3c2b630599b4eb7f70d755022a204905be63981","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-24 08:04:26.000000000","message":"Patch Set 1: Code-Review-1\n\nWe need to get a reno prelude and a couple of docs adjustments merged before this happens, if possible. That\u0027s on today\u0027s agenda.","accounts_in_message":[],"_revision_number":1},{"id":"38c15a84bc08a9d791393f93c09803ca8b1864f2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-24 08:33:01.000000000","message":"Patch Set 1:\n\nYes you still have the week... This is more of a placeholder. The commit message could be clearer...\n\nNo idea how reviewers get added, JP can probably elaborate.","accounts_in_message":[],"_revision_number":1},{"id":"045abdd6ec359f7ae8960c1231b545c9baf5a320","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-24 09:01:24.000000000","message":"Patch Set 1:\n\n\u003e Yes you still have the week... This is more of a placeholder. The\n \u003e commit message could be clearer...\n \u003e \n \u003e No idea how reviewers get added, JP can probably elaborate.\n\nMy script takes contact info from governance and from the liaisons defined in releases. It excludes any placeholder email like example@example.org.\n\nScript is good. What\u0027s not good is yours truly :) \nPebkac: I used master of governance as reference for containing the PTLs instead of a hash before the general PTL election results were validated. And because placement is a little later than others, the new PTL was not yet officially appointed at the time of the script run. So placement contained a placeholder email, which was removed from the set of emails to contact.","accounts_in_message":[],"_revision_number":1},{"id":"460faf384003874893be1149ac5509d5bcc7407a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-24 12:47:28.000000000","message":"Patch Set 1:\n\n\u003e the new PTL was not yet officially appointed\n\nFwiw, I think the old (train) ptl should be the one responsible for the train release patch anyway.","accounts_in_message":[],"_revision_number":1},{"id":"4e38f5cda951a7cdc46350d490cf0a38d2ce4ae6","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-24 16:14:01.000000000","message":"Patch Set 1:\n\n\u003e \u003e the new PTL was not yet officially appointed\n \u003e \n \u003e Fwiw, I think the old (train) ptl should be the one responsible for\n \u003e the train release patch anyway.\n\nAs said above, without pebkac the governance repo would be at the right sha, so you would get the right data :) Ahem ahem.","accounts_in_message":[],"_revision_number":1},{"id":"97a5ceb3db0eaed83099a465f932e254308e3aab","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-25 19:44:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"006e922df1463519ffb534d095e218a51e5deab1","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-25 19:44:40.000000000","message":"Patch Set 2:\n\nI\u0027ve updated the hash to point to the current head of master, which has the doc changes we wanted to see in train docs.","accounts_in_message":[],"_revision_number":2},{"id":"55d6f1b1ae591bda10e9f873c4c1acc0548ddeec","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-25 19:48:27.000000000","message":"Patch Set 2:\n\nGit hash looks good to me:\n\nhttps://opendev.org/openstack/placement/commit/2bf7ad034eead9328df1688adbeb8d415d6a7627","accounts_in_message":[],"_revision_number":2},{"id":"91b000ce4283e1ae8d415665b210594024e3f03e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-25 19:48:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b0ca9a9a27ca8854c923f497eaf72b70fe18a228","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-25 19:53:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"05cf1816fc8c5d26fd503bffd4267fc650a256af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 20:00:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd6784bf5284dc0b3a4de9b3ba17b93 : SUCCESS in 10m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef04f9cf021d492e8cb5e122b6b6a24d : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5a4ee68121974a6888e83a2d2c4bd9f5 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"42b8804144be8b77331fe44dc154032d2643d605","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-09-25 22:41:54.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good! Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"50da0c0143e0e22006f45512244de45ed7922461","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-09-26 05:40:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"20f16a6cd22c56b2032e5e66b1bca1ba8a3d2433","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-26 09:06:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b70a107ddf2172fd6b112308a61fc57a00a2904b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:07:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75cf74c04997eff19caed3a1b9da0c887600dcf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:34:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e26fcad46cf4ef6af98b7b95645805a : SUCCESS in 9m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30f45be7fc744a0e9e2eec92502d54e1 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b19ee905d4d1ad2d4fe010e4000e8c3f8986af85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:34:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"328a55e471bb6fda90854546b34f0fe26c756b17","revisions":{"0576b05a5c6837e005f4efba05eaa08691628e28":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:33:49.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/88/684088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/684088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/684088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/684088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/684088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/684088/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:33:42.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:33:42.000000000","tz":120},"subject":"Auto release request for placement","message":"Auto release request for placement\n\nChange-Id: I7a0945bfd2a91eee4e105fc38235ae6784289bc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0576b05a5c6837e005f4efba05eaa08691628e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0576b05a5c6837e005f4efba05eaa08691628e28"}]},"branch":"refs/heads/master"},"328a55e471bb6fda90854546b34f0fe26c756b17":{"kind":"REWORK","_number":2,"created":"2019-09-25 19:44:12.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/88/684088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/684088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/684088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/684088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/684088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/684088/2"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:33:42.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-25 19:44:05.000000000","tz":0},"subject":"Auto release request for placement","message":"Auto release request for placement\n\nChange-Id: I7a0945bfd2a91eee4e105fc38235ae6784289bc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/328a55e471bb6fda90854546b34f0fe26c756b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/328a55e471bb6fda90854546b34f0fe26c756b17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
