)]}'
{"id":"openstack%2Freleases~684102","triplet_id":"openstack%2Freleases~master~I3e0be9f02d69b04e2762a670d9b78e6b71a05870","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I3e0be9f02d69b04e2762a670d9b78e6b71a05870","subject":"Auto release request for searchlight-ui","status":"MERGED","created":"2019-09-23 18:38:36.000000000","updated":"2019-09-25 08:38:14.000000000","submitted":"2019-09-25 08:38:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684102-1569400694138-5622d755","meta_rev_id":"a649c40968e5e29ff6466bb066bb8af2725aae44","_number":684102,"virtual_id_number":684102,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 08:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-09-25 08:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-09-24 02:25:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-09-25 08:14:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 02:25:01.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-09-25 08:14:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-25 08:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13054d50647050b36172b950fdf3da7c32ce3b3f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:38:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"715a29dd8f4f8477a1e8fe086fb56da4fc78f809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 19:11:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdebec6ce28f4b15b5a7bdd1efe064c1 : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7fbc5d4dace34efaba9b57d607918c66 : FAILURE in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d8bcd58da3d34098bbe857d050591418 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9348553c8ab5c21762a7e9af822da46f49c84550","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-09-24 01:55:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef817d503ad6f34ca9ca2fea49c012d66c6c418f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 02:06:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d00d380f10479abe5926f7ffe3230d : SUCCESS in 9m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/734e124b34f34d61a14be50c4dca2e6c : SUCCESS in 6m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d2c21e9bcf47490abce8c9a8c8abc232 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"42e564e9f441800a797ab61f44f0233104ebe08c","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-09-24 02:25:01.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"3c63d858584a01295da9b4887fb9d20bf26839b7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-25 08:14:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dec543d1131325988f4d1455cf2ba5d201a28c09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 08:15:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e0377fde7e11fe9aa69967eeee8045fe1b82db9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 08:38:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40064e249a494c299099f9864db8d7eb : SUCCESS in 9m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a9bb37ec01b44cda0d92358dec0eff1 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8b120b4c7b1d3a019881b1dae83aedfee6279508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 08:38:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51dd2938fb07945f1191f405ddda02a0b56c6193","revisions":{"51dd2938fb07945f1191f405ddda02a0b56c6193":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:38:36.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/02/684102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/684102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/684102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/684102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/684102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/684102/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:38:29.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:38:29.000000000","tz":120},"subject":"Auto release request for searchlight-ui","message":"Auto release request for searchlight-ui\n\nChange-Id: I3e0be9f02d69b04e2762a670d9b78e6b71a05870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51dd2938fb07945f1191f405ddda02a0b56c6193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51dd2938fb07945f1191f405ddda02a0b56c6193"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
