)]}'
{"id":"openstack%2Freleases~684107","triplet_id":"openstack%2Freleases~master~I3c3e0f798895420eb0b6e76ffb86695c48c59e34","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"I3c3e0f798895420eb0b6e76ffb86695c48c59e34","subject":"Auto release request for storlets","status":"ABANDONED","created":"2019-09-23 18:40:40.000000000","updated":"2019-09-25 13:36:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dafb66ac9406709efae858bafade8222cf74940","_number":684107,"virtual_id_number":684107,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-09-23 19:17:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"all":[{"value":-1,"date":"2019-09-25 07:27:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 19:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-25 07:27:18.000000000","updated_by":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"reviewer":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"state":"REVIEWER"}],"messages":[{"id":"039822a44f54bfb9e9ec4c01bc7cf92798ca6539","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-23 18:40:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1baa0a2f6c1a13f42485fb8f27c72f8b8dc3288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 19:17:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67aa01284e14d08a3396b97c368bed2 : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98be6e9361bd4ce5902eb4b87476a8b5 : SUCCESS in 6m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/595140e217f1494eba28a8e0ec0eba7f : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fdd952062e6f41422e85efb04b40ee00c70db40f","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-09-25 07:27:18.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m still working for the storlets release. I\u0027ll make a patch for release today so please hold this unless I would not make any progress on upstream.","accounts_in_message":[],"_revision_number":1},{"id":"40ab468b894e0ef4440cee45913f7be624f49336","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-09-25 12:36:13.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/684733 is my proposal instead of this.","accounts_in_message":[],"_revision_number":1},{"id":"b7351a53bc9806b8d7aaed127dfe53ea144527c7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-25 13:36:33.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/#/c/684733/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d58baac3e2c86bec02fc3a3106c774d8f16ce0ac","revisions":{"d58baac3e2c86bec02fc3a3106c774d8f16ce0ac":{"kind":"REWORK","_number":1,"created":"2019-09-23 18:40:40.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/07/684107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/684107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/684107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/684107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/684107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/684107/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:40:33.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:40:33.000000000","tz":120},"subject":"Auto release request for storlets","message":"Auto release request for storlets\n\nChange-Id: I3c3e0f798895420eb0b6e76ffb86695c48c59e34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d58baac3e2c86bec02fc3a3106c774d8f16ce0ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d58baac3e2c86bec02fc3a3106c774d8f16ce0ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
