)]}'
{"id":"openstack%2Freleases~684244","triplet_id":"openstack%2Freleases~master~Ic88a981ab063f0a90114d0865fcb3d41333d3e7f","project":"openstack/releases","branch":"master","topic":"rc1-autorelease","hashtags":[],"change_id":"Ic88a981ab063f0a90114d0865fcb3d41333d3e7f","subject":"Auto release request for neutron-lbaas-dashboard","status":"ABANDONED","created":"2019-09-24 08:39:32.000000000","updated":"2019-09-24 12:03:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ee03a470278665799e4a8a8aff76c455d8f895b","_number":684244,"virtual_id_number":684244,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-24 09:12:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 09:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe1a95bcc1456eea267071b0c809caec47d849ab","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-09-24 08:39:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e348dea177ed3f262d016e0e83a77136827a8c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:12:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929d2c87de6a490da5a7b0e99638a99e : SUCCESS in 10m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66449b882c4d468194d9bb8497dbcb04 : FAILURE in 4m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4b4bf175d7eb4ba9a6463a86f39a50e5 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9d21ef3355d5e23a55b01e4001ea27f82a33c0f6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-24 12:03:33.000000000","message":"Abandoned\n\nneutron-lbaas is no longer a thing. See https://review.opendev.org/684291 instead","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9a5584e1e696efc5d0a3cda3c23332ad11dc8e5","revisions":{"e9a5584e1e696efc5d0a3cda3c23332ad11dc8e5":{"kind":"REWORK","_number":1,"created":"2019-09-24 08:39:32.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/44/684244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/684244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/684244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/684244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/684244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/684244/1"}}},"commit":{"parents":[{"commit":"4cb91b01ec11dda84b67f021e32223db18d83e81","subject":"Merge \"Cyborg cycle highlights for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cb91b01ec11dda84b67f021e32223db18d83e81"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:25:49.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-09-23 18:25:49.000000000","tz":120},"subject":"Auto release request for neutron-lbaas-dashboard","message":"Auto release request for neutron-lbaas-dashboard\n\nChange-Id: Ic88a981ab063f0a90114d0865fcb3d41333d3e7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9a5584e1e696efc5d0a3cda3c23332ad11dc8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9a5584e1e696efc5d0a3cda3c23332ad11dc8e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
