)]}'
{"id":"openstack%2Freleases~685050","triplet_id":"openstack%2Freleases~master~I014372c6d3e8a236fcc521c7cd818efe1eef4ea7","project":"openstack/releases","branch":"master","topic":"glance_store/1.0.0","hashtags":[],"change_id":"I014372c6d3e8a236fcc521c7cd818efe1eef4ea7","subject":"Release glance_store 1.0.1","status":"MERGED","created":"2019-09-26 12:02:00.000000000","updated":"2019-09-30 14:33:08.000000000","submitted":"2019-09-30 14:33:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685050-1569853988917-953ecec0","meta_rev_id":"d6cad3f205b6e99ced92c08dc844c97a1e5c09b8","_number":685050,"virtual_id_number":685050,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-30 14:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-30 14:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-09-30 14:14:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-30 14:18:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 16:04:22.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-27 19:23:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-30 14:14:59.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-09-30 14:18:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-30 14:33:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6adf11024ca7eabefe7908c7fb947b2b2821a85","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-26 12:02:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f563d42025e8d2709de1c7db4503958e238515a8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-26 12:02:29.000000000","message":"Patch Set 1:\n\nHold until, https://review.opendev.org/#/c/685035/ merges","accounts_in_message":[],"_revision_number":1},{"id":"a193156a4c921c33f0d6ce6f8e7cfe721d244162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:13:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51017bea69ec4554a869c605dc0c2278 : FAILURE in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/788bdf635b614c5591e444a1cb3c5154 : FAILURE in 4m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/92b5cb7330fc4a1287cee1235e954508 : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2b284ac20662e62ef987b4819e6806133351774f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-26 12:35:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ecb19879eb472eb1f4aa253b1a5fe9080ad1951b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:50:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80819cc7c81549c2a93546559778aea1 : SUCCESS in 9m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/630c4bbf6b4e484381dca3343b156c1e : FAILURE in 6m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f9501c29d9254d629fa3e2fc7c8f0be3 : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"fa76672671e2a97e2847e26bc2006ad1779e2723","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-26 14:26:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c28caa0688d8eb619d87bb68ff81314e83fc0996","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-26 14:37:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"754ae251131bf7bb1880eaca4359605cab112b07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-26 14:37:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"793d485e4ea30603790a363e33b4419805bdfe0b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-26 14:38:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ed7576a8bfe353b13540f2fa74e75b918f493ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:58:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c1542253133431e9efb6eba2d983a37 : SUCCESS in 9m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fb3d8eead784a828d311920b0fa5a18 : FAILURE in 5m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3f8126ece902427aab583029e0c6ca3d : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"76abecd8cb2db2537b8e508af05d36b14a4285fc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-26 16:04:22.000000000","message":"Patch Set 3:\n\nYou need to backport the fix to the stable/train branch if you really want a new release for it.\n\nWe are also under requirements freeze so you should open a thread on the ML to discuss the necessary requirements freeze exception.","accounts_in_message":[],"_revision_number":3},{"id":"b35fc75653e71f4ac73a050c0882f0153b450349","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 14:44:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c7f23559fc27ad16c7066a6a3bf4e62d789553b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-27 19:23:40.000000000","message":"Patch Set 3: Code-Review+1\n\nffe approved (once the sha is updated)","accounts_in_message":[],"_revision_number":3},{"id":"6c9cf13da2a5458f70cb316c7857e69d5f92d503","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-30 05:25:10.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"364355b7a090ac46509a126b1db1525be2f3b8af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 05:35:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c5c103327d499cb38d24b7e5915c2e : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9a46c52125e44498997433700663695 : SUCCESS in 4m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/88d01fbac1ac476f91e68eb3ce079fc1 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":4},{"id":"e1096da92a62f5a37e4771c12e683736534520f5","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-30 14:14:59.000000000","message":"Patch Set 4: Code-Review+1\n\nYes please, tag when possible. Hash is correct\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-September/009795.html","accounts_in_message":[],"_revision_number":4},{"id":"07e5f5c05837ff914b9c3ac7f6d9fc270f6e4d45","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-30 14:18:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nFFE has been approved and hash has been updated to pick up the desired commit. Pushing through to get this out right away.","accounts_in_message":[],"_revision_number":4},{"id":"c2f4ff9ddb87b6833fa19c926a366f165fe7f737","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:18:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db1dde384853a4d3d59014fd02ddd5cb50a203a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:33:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d74d39aaff541beaeb91446f6823d35 : SUCCESS in 9m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e87db3036a445568b81166045d83417 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":4},{"id":"cc0e133efd3319ce3019cd5dcfcdb60fbe345daa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:33:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4eb73374ef5f11fffdd2afa95c6f032b8e7f048","revisions":{"92fbd63c5ef4cdd600418bc855322ada044bfc65":{"kind":"REWORK","_number":1,"created":"2019-09-26 12:02:00.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/50/685050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/685050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/685050/1"}}},"commit":{"parents":[{"commit":"f46516a33642110dfd42f1d2660fc330dc539790","subject":"Merge \"Auto release request for networking-ovn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f46516a33642110dfd42f1d2660fc330dc539790"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 12:01:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 12:01:59.000000000","tz":0},"subject":"Release glance_store 1.0.1","message":"Release glance_store 1.0.1\n\nThis is the official release of glance_store version 1.0.1\n\nChange-Id: I014372c6d3e8a236fcc521c7cd818efe1eef4ea7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92fbd63c5ef4cdd600418bc855322ada044bfc65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92fbd63c5ef4cdd600418bc855322ada044bfc65"}]},"branch":"refs/heads/master"},"5cc7e9c07735b8e8322d995a3dc4895629b0f31c":{"kind":"REWORK","_number":2,"created":"2019-09-26 12:35:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/50/685050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/685050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/685050/2"}}},"commit":{"parents":[{"commit":"f46516a33642110dfd42f1d2660fc330dc539790","subject":"Merge \"Auto release request for networking-ovn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f46516a33642110dfd42f1d2660fc330dc539790"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 12:01:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 12:35:08.000000000","tz":0},"subject":"Release glance_store 1.0.1","message":"Release glance_store 1.0.1\n\nThis is the official release of glance_store version 1.0.1\n\nChange-Id: I014372c6d3e8a236fcc521c7cd818efe1eef4ea7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cc7e9c07735b8e8322d995a3dc4895629b0f31c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cc7e9c07735b8e8322d995a3dc4895629b0f31c"}]},"branch":"refs/heads/master"},"7a8866cf3fa2b41f23846bc84df0c2181e37eff2":{"kind":"REWORK","_number":3,"created":"2019-09-26 14:37:53.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/50/685050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/685050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/685050/3"}}},"commit":{"parents":[{"commit":"f46516a33642110dfd42f1d2660fc330dc539790","subject":"Merge \"Auto release request for networking-ovn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f46516a33642110dfd42f1d2660fc330dc539790"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 12:01:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 14:37:25.000000000","tz":0},"subject":"Release glance_store 1.0.1","message":"Release glance_store 1.0.1\n\nThis is the official release of glance_store version 1.0.1\n\nChange-Id: I014372c6d3e8a236fcc521c7cd818efe1eef4ea7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a8866cf3fa2b41f23846bc84df0c2181e37eff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a8866cf3fa2b41f23846bc84df0c2181e37eff2"}]},"branch":"refs/heads/master"},"a4eb73374ef5f11fffdd2afa95c6f032b8e7f048":{"kind":"REWORK","_number":4,"created":"2019-09-30 05:25:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/50/685050/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/685050/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/685050/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/685050/4"}}},"commit":{"parents":[{"commit":"f46516a33642110dfd42f1d2660fc330dc539790","subject":"Merge \"Auto release request for networking-ovn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f46516a33642110dfd42f1d2660fc330dc539790"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-26 12:01:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-09-30 05:25:01.000000000","tz":0},"subject":"Release glance_store 1.0.1","message":"Release glance_store 1.0.1\n\nThis is the official release of glance_store version 1.0.1\n\nChange-Id: I014372c6d3e8a236fcc521c7cd818efe1eef4ea7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4eb73374ef5f11fffdd2afa95c6f032b8e7f048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4eb73374ef5f11fffdd2afa95c6f032b8e7f048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
