)]}'
{"id":"openstack%2Freleases~685067","triplet_id":"openstack%2Freleases~master~Ia29828e26a8a17af8d7e50906e5340c1327e8c2a","project":"openstack/releases","branch":"master","topic":"ironic-ui-train","hashtags":[],"change_id":"Ia29828e26a8a17af8d7e50906e5340c1327e8c2a","subject":"Release ironic-ui 3.5.2 and create stable/train","status":"MERGED","created":"2019-09-26 13:06:10.000000000","updated":"2019-09-26 19:47:58.000000000","submitted":"2019-09-26 19:47:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685067-1569527278041-dadc74e4","meta_rev_id":"8dd6f8fb3cd8c5f99d15f2c36df6bfe111a0ebb8","_number":685067,"virtual_id_number":685067,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 19:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-26 19:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-26 16:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-26 15:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-09-26 19:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 15:17:57.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-09-26 16:01:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-09-26 19:31:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-26 19:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33a93570298ea3a8d012c0863e3884cab29c6367","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 13:06:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f49486e0f982e0f256055acc2d19b8ce311a0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 13:23:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6dee65784d4c9f8458ef9323c49874 : SUCCESS in 9m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ed5d31a27464222bd91ea089e94e4c7 : SUCCESS in 5m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d2b3da2fbad845f8b97217796053aa08 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6b8d340c7b90e6a744566594a8ef29555890dba1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-26 13:43:10.000000000","message":"Patch Set 1:\n\nThere is another commit to build PDF docs, don\u0027t you want to include it?","accounts_in_message":[],"_revision_number":1},{"id":"ed3fd4c1c527c01d7f8a394ebd4c333e2d38964c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 13:44:19.000000000","message":"Patch Set 1:\n\n\u003e There is another commit to build PDF docs, don\u0027t you want to\n \u003e include it?\n\nDidn\u0027t think it was worth releasing for that, given that it\u0027s not a code change. Disagree?","accounts_in_message":[],"_revision_number":1},{"id":"3aaa7a7551e2f9e0c38bd22bd0d771d6fda217cd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-26 13:45:23.000000000","message":"Patch Set 1:\n\nBuilding PDF docs was a Train goal, so ideally we should include it.","accounts_in_message":[],"_revision_number":1},{"id":"b6cb0d219b26cac96fcec104fc4d7b941a9cd0f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-26 14:35:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d24366af50d5e6f12475e4e22d9c524ffb55311a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 14:53:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a3695c7f534e6ba72400a0f522282e : SUCCESS in 10m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e06b32b1a7fe453495b67b2f94a3f263 : SUCCESS in 5m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4e176038cc06468f84e1bde492127c53 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"39b871debde0aa94773904f9fa07f90d102d5993","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-26 15:17:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2097c315abbcd28cfee8187b833cef1e5fbbe36e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-26 16:01:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a77db990a47ca1ffee54bb1aa336ec9e40d279d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-26 19:31:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a73591fee4fbee56c2fa7284c598106e2d16a679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:32:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa5205b1561673abf160f449bb3fbded1256b2d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:47:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8acc0941b1b041078ecdec17ba17e790 : SUCCESS in 9m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15e8d65a0d6e4140bfeb636a560ad2b5 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e4f3ed0f3f3d489fa609b214239b25315b92947e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 19:47:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab1102c892d4a68c09734e57dae948da8d67df76","revisions":{"c2167e61820d17ae8c351d8c847016a0afa63706":{"kind":"REWORK","_number":1,"created":"2019-09-26 13:06:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/685067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/685067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/685067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/685067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/685067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/685067/1"}}},"commit":{"parents":[{"commit":"487cb58c8e58bb7ab720c8ea763b1303335ecb64","subject":"Merge \"Auto release request for monasca-kibana-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/487cb58c8e58bb7ab720c8ea763b1303335ecb64"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-26 13:05:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-26 13:05:44.000000000","tz":60},"subject":"Create stable/train for ironic-ui","message":"Create stable/train for ironic-ui\n\nChange-Id: Ia29828e26a8a17af8d7e50906e5340c1327e8c2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c2167e61820d17ae8c351d8c847016a0afa63706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c2167e61820d17ae8c351d8c847016a0afa63706"}]},"branch":"refs/heads/master"},"ab1102c892d4a68c09734e57dae948da8d67df76":{"kind":"REWORK","_number":2,"created":"2019-09-26 14:35:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/685067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/685067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/685067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/685067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/685067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/685067/2"}}},"commit":{"parents":[{"commit":"487cb58c8e58bb7ab720c8ea763b1303335ecb64","subject":"Merge \"Auto release request for monasca-kibana-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/487cb58c8e58bb7ab720c8ea763b1303335ecb64"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-26 13:05:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-26 14:35:22.000000000","tz":60},"subject":"Release ironic-ui 3.5.2 and create stable/train","message":"Release ironic-ui 3.5.2 and create stable/train\n\nChange-Id: Ia29828e26a8a17af8d7e50906e5340c1327e8c2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab1102c892d4a68c09734e57dae948da8d67df76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab1102c892d4a68c09734e57dae948da8d67df76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
