)]}'
{"id":"openstack%2Freleases~685305","triplet_id":"openstack%2Freleases~master~I3372457effee17c4392a2c50aba398360ba31cd3","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I3372457effee17c4392a2c50aba398360ba31cd3","subject":"release Tricircle for Trian","status":"ABANDONED","created":"2019-09-27 09:54:40.000000000","updated":"2019-09-27 15:31:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae487e84ffd153a5d9bd763d68306ee91a7a68ff","_number":685305,"virtual_id_number":685305,"owner":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-27 15:23:25.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-27 10:09:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2019-09-27 15:31:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 10:09:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-27 15:23:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-27 15:31:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"e0c47f36a8f472de14df8703be70b6b4bc5757be","author":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"date":"2019-09-27 09:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f16a9acb077db13986d535266091e88455771d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:09:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d55ac0e9fa488790010755f870702b : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce97aab627ed4000895a7ea2abf06977 : FAILURE in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1da4ece88784ac0bae4c794b006b689 : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dc8ddb1a884ab4f207764f395efeb5105aebe0cc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 11:58:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6289d6ccce0a52a9438f92f89c1635dc7b33ea8c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 12:02:21.000000000","message":"Patch Set 1:\n\nAlso this needs to propose a RC. See https://review.opendev.org/#/c/684109/ -- which you can alter with a more appropriate SHA if you want.","accounts_in_message":[],"_revision_number":1},{"id":"1f80d52a103ea682b3706b677c83dc378937181b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 14:40:31.000000000","message":"Patch Set 1:\n\nThis may be useful for you in the future:\n\nhttps://releases.openstack.org/reference/using.html#using-new-release-command","accounts_in_message":[],"_revision_number":1},{"id":"3bd359b5943f302e2a22774ff487ec194c91aa57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-27 15:23:25.000000000","message":"Patch Set 1:\n\nAlso, may be better to just update the hash in https://review.opendev.org/684109","accounts_in_message":[],"_revision_number":1},{"id":"5d5a2c26fc8898142eb06e7a608c4f85dffcfc78","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 15:31:29.000000000","message":"Patch Set 1: Code-Review-1\n\nOK, so we will approve https://review.opendev.org/#/c/685305/ (which release tricircle RC1 with the current commit SHA for master) on Monday.\n\nIf you wanted to have https://review.opendev.org/#/c/684159/ in before, please get it merged between now and then, and comment on https://review.opendev.org/#/c/685305/.\n\nThis review will be abandoned since it is further away from a Tricircle RC1 than https://review.opendev.org/#/c/685305/","accounts_in_message":[],"_revision_number":1},{"id":"ae487e84ffd153a5d9bd763d68306ee91a7a68ff","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-27 15:31:39.000000000","message":"Abandoned\n\nSuperseded by https://review.opendev.org/#/c/685305/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96764e96c491cf6b55f48f93e7a27b214ed83c5b","revisions":{"96764e96c491cf6b55f48f93e7a27b214ed83c5b":{"kind":"REWORK","_number":1,"created":"2019-09-27 09:54:40.000000000","uploader":{"_account_id":28071,"name":"chzhang8","email":"zhangchi@szzt.com.cn","username":"chzhang"},"ref":"refs/changes/05/685305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/685305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/685305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/685305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/685305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/685305/1"}}},"commit":{"parents":[{"commit":"fdac3e4aacefb7e593404121afe51ad9fcb654c3","subject":"Merge \"Auto release request for monasca-notification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fdac3e4aacefb7e593404121afe51ad9fcb654c3"}]}],"author":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2019-09-27 09:52:50.000000000","tz":480},"committer":{"name":"zhangchi","email":"zhangchi@szzt.com.cn","date":"2019-09-27 09:54:37.000000000","tz":480},"subject":"release Tricircle for Trian","message":"release Tricircle for Trian\n\nChange-Id: I3372457effee17c4392a2c50aba398360ba31cd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96764e96c491cf6b55f48f93e7a27b214ed83c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/96764e96c491cf6b55f48f93e7a27b214ed83c5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
