)]}'
{"id":"openstack%2Freleases~686069","triplet_id":"openstack%2Freleases~master~I58f589d721660c6ca9d9464bdb0222a837b7b815","project":"openstack/releases","branch":"master","topic":"cinderproject-aug-2019","hashtags":[],"change_id":"I58f589d721660c6ca9d9464bdb0222a837b7b815","subject":"Release cinder 12.0.9","status":"MERGED","created":"2019-10-01 22:39:44.000000000","updated":"2019-10-04 12:56:22.000000000","submitted":"2019-10-04 12:56:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686069-1570193782447-01385176","meta_rev_id":"6c780b331bd2b25a8f1b4db9234cc6f90ec30f7b","_number":686069,"virtual_id_number":686069,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 12:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-10-04 04:57:02.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-10-02 09:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-10-04 12:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-10-04 12:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 09:32:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-04 04:57:02.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-10-04 12:40:43.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-04 12:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfe2fff06d7c49c9c9ce22284589e4b0710c2e2e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 22:39:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dd3a5d81346a0326c15c8c49aa805f69a29ea7c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 22:41:54.000000000","message":"Patch Set 1: Code-Review-1\n\nIncorrect hash.","accounts_in_message":[],"_revision_number":1},{"id":"cfa85771b6f48916170db34838ea70d8e907f963","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 22:45:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fae10f1cb4e6bf2a39340c2566a4db9c2083387","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 22:49:14.000000000","message":"Topic changed from release-cinder-12.0.9 to cinderproject-aug-2019","accounts_in_message":[],"_revision_number":2},{"id":"b30321c3c1ec882aef9b6815d353088106de1c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 22:54:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d3d411c6df46e285240b315eaabbd3 : SUCCESS in 7m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05a032a7ace6492a8072224b93b3493f : SUCCESS in 6m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f610bd0892954a318d88777238111e85 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f46074c248ef69ce534a30c0840897fdf493a99a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-02 09:32:26.000000000","message":"Patch Set 2: Code-Review+2\n\nNeeds stable review but the commits look like bugfixes to me","accounts_in_message":[],"_revision_number":2},{"id":"36cf6ed97c5d05ff46bcb429ea6e68e6aa3d163a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-02 11:07:41.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM but someone from stable should approve.","accounts_in_message":[],"_revision_number":2},{"id":"eeeb15618c63b213a67a83857d44aad57cce7892","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-10-04 04:57:02.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me.\n\nNot approving as I\u0027ve been AWOL and I don\u0027t want to mess things up on a Friday.","accounts_in_message":[],"_revision_number":2},{"id":"70bc91eb8a1c9d7fda8ded77cb21d39185c5b828","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:40:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm, signed off by stable reviewer too.","accounts_in_message":[],"_revision_number":2},{"id":"1fa311d2ef8e0896465b78d62749b26644f494bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 12:40:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d88860eef65d4a28cd6c3f5bc4c0e2d024da833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 12:56:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/655cfdf3398649ea9f57a5ceadcab6d0 : SUCCESS in 11m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e13f31d611904ce5aaa3b0fae528cc43 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6c780b331bd2b25a8f1b4db9234cc6f90ec30f7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 12:56:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"458ca53f18bda295481990d27842372e7b097b2c","revisions":{"98a0d7109fe6c90839b150d4ec2b540da84eae95":{"kind":"REWORK","_number":1,"created":"2019-10-01 22:39:44.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/69/686069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/686069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/686069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/686069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/686069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/686069/1"}}},"commit":{"parents":[{"commit":"4ccb8067787305a1ffb162d1922b05308c99003b","subject":"Merge \"Release 1.0.0 for ironic-python-agent-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ccb8067787305a1ffb162d1922b05308c99003b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-10-01 22:39:02.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-10-01 22:39:02.000000000","tz":-240},"subject":"Release cinder 12.0.9","message":"Release cinder 12.0.9\n\nRelease from stable/queens\n\nChange-Id: I58f589d721660c6ca9d9464bdb0222a837b7b815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98a0d7109fe6c90839b150d4ec2b540da84eae95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98a0d7109fe6c90839b150d4ec2b540da84eae95"}]},"branch":"refs/heads/master"},"458ca53f18bda295481990d27842372e7b097b2c":{"kind":"REWORK","_number":2,"created":"2019-10-01 22:45:04.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/69/686069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/686069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/686069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/686069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/686069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/686069/2"}}},"commit":{"parents":[{"commit":"4ccb8067787305a1ffb162d1922b05308c99003b","subject":"Merge \"Release 1.0.0 for ironic-python-agent-builder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ccb8067787305a1ffb162d1922b05308c99003b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-10-01 22:39:02.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-10-01 22:44:55.000000000","tz":-240},"subject":"Release cinder 12.0.9","message":"Release cinder 12.0.9\n\nRelease from stable/queens\n\nChange-Id: I58f589d721660c6ca9d9464bdb0222a837b7b815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/458ca53f18bda295481990d27842372e7b097b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/458ca53f18bda295481990d27842372e7b097b2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
