)]}'
{"id":"openstack%2Freleases~686303","triplet_id":"openstack%2Freleases~master~Iab0a45eb4322dec3c047fba1662fcd36ab255436","project":"openstack/releases","branch":"master","topic":"manila","hashtags":[],"change_id":"Iab0a45eb4322dec3c047fba1662fcd36ab255436","subject":"Add Ussuri cycle schedule for manila","status":"MERGED","created":"2019-10-02 21:34:21.000000000","updated":"2019-10-08 15:01:47.000000000","submitted":"2019-10-08 15:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686303-1570546907245-13ad88d9","meta_rev_id":"726438ebad9b0c1d864aef5ba37c4099e6a99446","_number":686303,"virtual_id_number":686303,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 15:01:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-08 13:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-07 23:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2019-10-08 10:06:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-08 14:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-08 14:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 23:31:35.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2019-10-08 10:06:16.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-10-08 13:39:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-08 14:51:33.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-08 15:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"618b61e9abeb53463f4f6a34977483eeab11ae89","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-02 21:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"951296108a90b574609a042fa3ffc737941c7cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 21:48:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5405bc739e84d469cb9afa94be47685 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f93e3dfea70f9cde7cf66305a6beb7370dd2258a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-02 22:50:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0820872709dcae30a358e82eb64ede148471819a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-10-03 13:44:24.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nGood to have u-manila-fpfreeze earlier again. Thanks Goutham.","accounts_in_message":[],"_revision_number":1},{"id":"2c79ba0eeba2b555441e2c368ada9abb7e139afa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-07 14:23:55.000000000","message":"Patch Set 1:\n\nLooks fine from the release schedule perspective, but I\u0027ll wait for the team to come to agreement on what they want it to say.","accounts_in_message":[],"_revision_number":1},{"id":"de6aff76649676533bd4fbb492548ef71dba2def","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-07 21:00:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"120d254666fbf50d1a358565836ea7dd4d1c3354","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-07 21:36:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6adbfe7cb55f6f4aee8df8bd586a605c8e83b29e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 21:50:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3114c3d6198a4a4caba5ffd4615b7fd7 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"237e376be648afab590f518ca7317dde79a1ac98","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-10-07 23:31:35.000000000","message":"Patch Set 2: Code-Review+1\n\nGoutham, thanks for clarifying this. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"7baaf47af094d4ca3df39fb2c6f8a35d52e51f18","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-08 10:06:16.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks for the clarification, Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"f7b101efa37fd0b2e00a10f989317f789326c868","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-08 13:39:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6d98518928aa993bfce354f0b07059ce7d78945","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-08 14:51:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nIf it\u0027s fine for you, it\u0027s fine for me!","accounts_in_message":[],"_revision_number":2},{"id":"87f768331d7e000bfdab540ef59bdab21bbb2b23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 14:51:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9eed90a56cec881b9eea895f74daef1123162a50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:01:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cf6a5533e03d6c284b59adee0659ca5a78bafc90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:01:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c93c0db1622452b9b8caa692c5cda9f : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14573c7b24703ac5aa9ffb226ccedad15c1dbf25","revisions":{"4e6658d54f580c820498eea460300e0089d6fb84":{"kind":"REWORK","_number":1,"created":"2019-10-02 21:34:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/03/686303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/686303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/686303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/686303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/686303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/686303/1"}}},"commit":{"parents":[{"commit":"5d7e4f9ac355b8083a13ac63fab8fd1e119b11c2","subject":"Merge \"Add stable/train for neutron-[fwaas|vpnaas]-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d7e4f9ac355b8083a13ac63fab8fd1e119b11c2"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-02 21:34:06.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-02 21:34:06.000000000","tz":-420},"subject":"Add Ussuri cycle schedule for manila","message":"Add Ussuri cycle schedule for manila\n\nChange-Id: Iab0a45eb4322dec3c047fba1662fcd36ab255436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e6658d54f580c820498eea460300e0089d6fb84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e6658d54f580c820498eea460300e0089d6fb84"}]},"branch":"refs/heads/master"},"14573c7b24703ac5aa9ffb226ccedad15c1dbf25":{"kind":"REWORK","_number":2,"created":"2019-10-07 21:36:45.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/03/686303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/686303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/686303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/686303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/686303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/686303/2"}}},"commit":{"parents":[{"commit":"5d7e4f9ac355b8083a13ac63fab8fd1e119b11c2","subject":"Merge \"Add stable/train for neutron-[fwaas|vpnaas]-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d7e4f9ac355b8083a13ac63fab8fd1e119b11c2"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-02 21:34:06.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-07 21:36:25.000000000","tz":-420},"subject":"Add Ussuri cycle schedule for manila","message":"Add Ussuri cycle schedule for manila\n\nChange-Id: Iab0a45eb4322dec3c047fba1662fcd36ab255436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14573c7b24703ac5aa9ffb226ccedad15c1dbf25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14573c7b24703ac5aa9ffb226ccedad15c1dbf25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
