)]}'
{"id":"openstack%2Freleases~686492","triplet_id":"openstack%2Freleases~master~I8507d86e4f553b698163be61665267d42e033d91","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I8507d86e4f553b698163be61665267d42e033d91","subject":"Redirects to master should always be temporary","status":"MERGED","created":"2019-10-03 21:01:43.000000000","updated":"2019-10-04 08:11:13.000000000","submitted":"2019-10-04 08:11:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686492-1570176673488-b42f53c0","meta_rev_id":"5a5b8e4b1a19c8d9d0476ae47016a2838adad9e8","_number":686492,"virtual_id_number":686492,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 08:11:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-10-04 04:39:52.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-04 07:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-04 07:59:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 04:39:52.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-10-04 07:59:55.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-04 08:11:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"203861e6136471c62ea7aab862bb4cc957c8234b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-03 21:01:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f008152d99dc3427a8846f06a55ff12eb9de9da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 21:17:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2433c212bc7c4493a0b41ce0079e8765 : FAILURE in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e7b4a1c099d48fc903dd05565244163 : FAILURE in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3efb4c01a8e54a239b543403488aa09a : SUCCESS in 9m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/855c96665d8e485b8b75ce249651b5fb : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6780cd9d1ac84c108195460df2de73f9 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdf82dbfe614ab2ba58b494c1cde5c2 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b1271062860425b5a86f546dbdcf40310559fad7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-03 22:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01693917e29f46181ec6c1e8e4d12c4471a1045f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 23:17:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ded92ee57ca4eaa9e26ce3b717c4db6 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51bab1f55fd34458a1362228dc951786 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ff41aa98b74c4f98ce117fca6d9c8d : SUCCESS in 8m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc07affa3de6405fa4e9605f0052d906 : SUCCESS in 4m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b7a3ef464e604f5cbed8de56f5930248 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c1901653f644eb87288cab3e16e367 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"4907c51262d37232bc897ad77483978cbfbcc240","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-10-04 04:39:52.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me.\n\nThanks Tim.","accounts_in_message":[],"_revision_number":2},{"id":"f64585914c9122468780579c86b0e6c827f3e7c4","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 07:59:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGood catch.","accounts_in_message":[],"_revision_number":2},{"id":"d0db3d152c88456b5d495aaa5b718b2943a191ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:00:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de15cdf85c57c67c6875b991f312b5c7b4da8d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:11:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e5a30b65e34a29b0607e54636c2a61 : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cea66f8ee25475b9205e3302a4a63d3 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47125064fa7042a08521bce28dd91a99 : SUCCESS in 9m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/496605e71b3345068f76471b691b204c : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b39732cf14acec69b33b21bb123e124ac140dc58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:11:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e392f0bd481d1b78f54d23faa8ce62894be1e3b9","revisions":{"6e7647b853f9e3fda49df0f121ef092819ed9653":{"kind":"REWORK","_number":1,"created":"2019-10-03 21:01:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/686492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/686492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/686492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/686492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/686492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/686492/1"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-03 20:58:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-03 20:58:21.000000000","tz":-420},"subject":"Redirects to master should always be temporary","message":"Redirects to master should always be temporary\n\nWe *know* that requirements will have a stable/train branch soon; make\nsure clients don\u0027t go caching the redirect to master.\n\nChange-Id: I8507d86e4f553b698163be61665267d42e033d91\nRelated-Change: Ia2e8c46c27ac97217576afdd1677efba4b99fc37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7647b853f9e3fda49df0f121ef092819ed9653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e7647b853f9e3fda49df0f121ef092819ed9653"}]},"branch":"refs/heads/master"},"e392f0bd481d1b78f54d23faa8ce62894be1e3b9":{"kind":"REWORK","_number":2,"created":"2019-10-03 22:10:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/686492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/686492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/686492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/686492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/686492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/686492/2"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-03 20:58:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-03 22:10:33.000000000","tz":-420},"subject":"Redirects to master should always be temporary","message":"Redirects to master should always be temporary\n\nWe *know* that requirements will have a stable/train branch soon; make\nsure clients don\u0027t go caching the redirect to master.\n\nChange-Id: I8507d86e4f553b698163be61665267d42e033d91\nRelated-Change: Ia2e8c46c27ac97217576afdd1677efba4b99fc37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e392f0bd481d1b78f54d23faa8ce62894be1e3b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e392f0bd481d1b78f54d23faa8ce62894be1e3b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
