)]}'
{"id":"openstack%2Freleases~686578","triplet_id":"openstack%2Freleases~master~I51bffb9109b6b874aef660932b7bfba87ac5c2ab","project":"openstack/releases","branch":"master","topic":"storlets-train-rc2","hashtags":[],"change_id":"I51bffb9109b6b874aef660932b7bfba87ac5c2ab","subject":"Release storlets rc2 for train","status":"MERGED","created":"2019-10-04 12:01:07.000000000","updated":"2019-10-07 11:44:17.000000000","submitted":"2019-10-07 11:44:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686578-1570448657225-fe29816b","meta_rev_id":"fb0c94105f9c47d942342984f4ac974c4c3b3725","_number":686578,"virtual_id_number":686578,"owner":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-04 20:26:09.000000000","_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 11:44:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":2,"date":"2019-10-06 16:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-07 11:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-07 11:33:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-06 16:57:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-07 11:33:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-07 11:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"730df07f864980a670b606867b518a05f4338321","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-10-04 12:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1fc85562aa5b09e43afe52de19a980dd9daeb6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 12:28:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd4e69ba0ef47f285aeef99dad02a09 : FAILURE in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3fc3c8827624c93ae60aeac14a6dcf9 : FAILURE in 3m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9624951f6c294df9954d90b6ee7ab3b0 : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9dfe698bf20061701d88fbe3576e0a88de64c59e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-04 12:48:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0a0b1c716036aa90e1172dbb5d57944554bbd9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:01:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4c67b047bb4274a2b44ce016cefdf2 : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f61572852114447a14a8e4f7f6d5e5f : FAILURE in 4m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7b888045d0c84de18645cad7abb45e40 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f5d2797af8885a57a04e7040e4fcfa7c60147c52","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-10-04 15:45:53.000000000","message":"Patch Set 2:\n\nit seems that the hash should be that in stable/train?\n\nI set the hash at the master, will try to change to the one at https://review.opendev.org/#/c/686577/","accounts_in_message":[],"_revision_number":2},{"id":"db69190c55faec9f7425bfcf025e82a87922ea48","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-10-04 15:51:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4780aa6bdabd5afe44538075b094ce9aff0517c4","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-10-04 15:52:22.000000000","message":"Patch Set 3:\n\nthe patch is still in the gate before landing but push the newest hash before I enter to asleep. I\u0027ll try to recheck tomorrow morning.","accounts_in_message":[],"_revision_number":3},{"id":"9dc588109eed1c53bb54f9621452c447307f4eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:07:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82641ca0ff08488b841673878debb18b : SUCCESS in 10m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f583f59fdf0549fcad894fb3d61fe853 : FAILURE in 7m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/00b28d04bfca4eba8729cdb319cce151 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":3},{"id":"bd7f61725f271a1be3ec4abad9968e19d05a6b72","author":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"date":"2019-10-04 20:26:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ea5e670fe75623f383bf69d4cfc3b802ec4bd851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 21:01:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b0546e19ec4187822d8e47af9b05f8 : SUCCESS in 10m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb66134940464c4faa220eacf7d73767 : SUCCESS in 5m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/013b31bbc6134872956cf227cabd1d56 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":3},{"id":"ae30bd02156888a5178a7f39adef318f7720cdfd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-06 16:57:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cbbcfbd68a3e803e2da59be146f3fd9044a6fe5d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-07 11:33:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"79e097118f64df660ef7e2f8e55fae95a045fafc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:33:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e9679cfc81bdd776d46b3b5491e7ec31108493f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:44:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"18eb43f7e75f81c6f931f795844d5368e73e4b44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:44:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d39b4594fe4cacb480fb6a4c7437e6 : SUCCESS in 9m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6efc235ae7c247a7bac27fc58aa8f4cb : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"271f3fed095fb1f987d0386e78574f876b1d5497","revisions":{"67718482d242b53a74d537ca1a6722457a805e05":{"kind":"REWORK","_number":1,"created":"2019-10-04 12:01:07.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/78/686578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/686578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/686578/1"}}},"commit":{"parents":[{"commit":"46ec6a4dc0f6b1583bc6d9c14320576455b988bc","subject":"Merge \"Add stable/train for tacker-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46ec6a4dc0f6b1583bc6d9c14320576455b988bc"}]}],"author":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-10-04 11:59:35.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-10-04 12:00:42.000000000","tz":540},"subject":"Release storlets rc2 for train","message":"Release storlets rc2 for train\n\nbecause rc1 was missing the release note for 4.0.0\nthe new hash for rc2 includes the release note change.\n\nChange-Id: I51bffb9109b6b874aef660932b7bfba87ac5c2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67718482d242b53a74d537ca1a6722457a805e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67718482d242b53a74d537ca1a6722457a805e05"}]},"branch":"refs/heads/master"},"63e44e77a03980381235ef72608a6ff72f464acb":{"kind":"REWORK","_number":2,"created":"2019-10-04 12:48:25.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/78/686578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/686578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/686578/2"}}},"commit":{"parents":[{"commit":"46ec6a4dc0f6b1583bc6d9c14320576455b988bc","subject":"Merge \"Add stable/train for tacker-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46ec6a4dc0f6b1583bc6d9c14320576455b988bc"}]}],"author":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-10-04 11:59:35.000000000","tz":540},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-10-04 12:48:20.000000000","tz":0},"subject":"Release storlets rc2 for train","message":"Release storlets rc2 for train\n\nbecause rc1 was missing the release note for 4.0.0\nthe new hash for rc2 includes the release note change.\n\nChange-Id: I51bffb9109b6b874aef660932b7bfba87ac5c2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63e44e77a03980381235ef72608a6ff72f464acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63e44e77a03980381235ef72608a6ff72f464acb"}]},"branch":"refs/heads/master"},"271f3fed095fb1f987d0386e78574f876b1d5497":{"kind":"REWORK","_number":3,"created":"2019-10-04 15:51:12.000000000","uploader":{"_account_id":4608,"name":"Kota Tsuyuzaki","email":"bloodeagle40234@gmail.com","username":"tsuyuzaki-kota"},"ref":"refs/changes/78/686578/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/686578/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/686578/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/686578/3"}}},"commit":{"parents":[{"commit":"46ec6a4dc0f6b1583bc6d9c14320576455b988bc","subject":"Merge \"Add stable/train for tacker-horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46ec6a4dc0f6b1583bc6d9c14320576455b988bc"}]}],"author":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-10-04 11:59:35.000000000","tz":540},"committer":{"name":"Kota Tsuyuzaki","email":"kota.tsuyuzaki.pc@hco.ntt.co.jp","date":"2019-10-04 15:50:49.000000000","tz":540},"subject":"Release storlets rc2 for train","message":"Release storlets rc2 for train\n\nbecause rc1 was missing the release note for 4.0.0\nthe new hash for rc2 includes the release note change.\n\nChange-Id: I51bffb9109b6b874aef660932b7bfba87ac5c2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/271f3fed095fb1f987d0386e78574f876b1d5497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/271f3fed095fb1f987d0386e78574f876b1d5497"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
