)]}'
{"id":"openstack%2Freleases~686581","triplet_id":"openstack%2Freleases~master~If34f26d604ff2e52cbbacc31a46f4a8aef6a0258","project":"openstack/releases","branch":"master","topic":"create-train","hashtags":[],"change_id":"If34f26d604ff2e52cbbacc31a46f4a8aef6a0258","subject":"Create train branch for OSA roles","status":"MERGED","created":"2019-10-04 12:16:34.000000000","updated":"2019-10-22 18:47:31.000000000","submitted":"2019-10-22 18:47:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686581-1571770051690-8b4ac2a1","meta_rev_id":"2bc72421763349bd1654d4233627201642eb5eb0","_number":686581,"virtual_id_number":686581,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 18:47:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-21 18:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-22 18:30:20.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-22 18:35:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 14:28:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-22 18:30:20.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-10-22 18:35:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-22 18:47:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca6a614187b420212f5496378fb7d8aef3cb3c70","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8bf11fc4a71ccf3ae5dac670c20429af20b92e5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:17:00.000000000","message":"Patch Set 1: Workflow-1\n\nI didn\u0027t get the chance to speak to mnaser about this, but I think we should do sooner rather than later.","accounts_in_message":[],"_revision_number":1},{"id":"996bdbde594daca427ee3ac45b940b85b63d6264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 12:23:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f1ed758744d4fa7ba5bf8bcd769618e : FAILURE in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93cae2c9204e4c5da68d069f9166e082 : FAILURE in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/775901eec3384af2bed20f3478a0116c : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b1c00d3e27fbb313f8bf75757206ecbf2ba32e17","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 14:25:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2524c284a3e77249cc5602b2c7ee8320c32a08e4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-21 14:26:33.000000000","message":"Patch Set 2:\n\nLooks like this needs branch name.","accounts_in_message":[],"_revision_number":2},{"id":"e4c73e3e0292fdf427d1136a244ffccb099dd74e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 14:28:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9eb47ea96397603f2767d72a5924a210abe6100d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 14:28:57.000000000","message":"Patch Set 2:\n\noh, yep, that\u0027s right, thanks for the hint!","accounts_in_message":[],"_revision_number":2},{"id":"188d733c04ef3f75bc170fd90d77d6b9f80629cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:57:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae94f977a6934ec1a7c0d71952874052 : SUCCESS in 10m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0f77b327567430682b98df46dd1a8b1 : FAILURE in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/06c5606981f5424ca91c62094e416731 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1347971a7f880ec42e2eb34753bf12aa5dfdfd46","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 15:21:31.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9622ff098b35370cbedbf8f43c0a5e1374600179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 15:35:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b66b75c782c43cc8a0c7afc887151a5 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5d8d8b8446f45bc967556c1f7dcbb87 : FAILURE in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4c0f8bd3f5df48309ba30b3d3a99a422 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"79f1e02edbc7e3462a7d37fe02fd18b49a8d44d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-21 15:47:06.000000000","message":"Patch Set 4:\n\ntox -e validate can be run locally to find any other missing deliverables.","accounts_in_message":[],"_revision_number":4},{"id":"7db2a637907e0f85d41e32f4db0555967b021831","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 17:48:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"972f6927e1fb1340bb1b30afb70ebf6ce899cc47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 18:02:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099315adf02e41e380d56211bc9be818 : SUCCESS in 8m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29b4f314b0524eb8902ff07b3f7dbc2e : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9fa6e8d4675748f5ab3269a76af7ab32 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":5},{"id":"284a97b0157e2460e05d7e30a140a856845420f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-21 18:07:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"98f8d81e23d66c4186ff9de55d0482b05bebfb92","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-10-22 18:30:20.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks good to me","accounts_in_message":[],"_revision_number":5},{"id":"d96fdb7487810d852767e4fddae391a39e80c3dd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 18:35:34.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d18de78dee6a4e2e26f445e87f827365d1440103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:35:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ba5c91b1682c68ce379a3c0924d54d5b8c027d30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:47:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"bf2c7a1102bad4c69463d9879b8b8da2768eb937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:47:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d2d2b04cd14273b1dd2087318583d9 : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8a52bd02591471387431f0168345d4e : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0086a2883d3cb6daf156078067e019acacd6496f","revisions":{"4eb787a644554b6aabe18b5052594f9fe392aa13":{"kind":"REWORK","_number":1,"created":"2019-10-04 12:16:34.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/81/686581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/686581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/686581/1"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:15:52.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:15:52.000000000","tz":120},"subject":"Create train branch for OSA roles","message":"Create train branch for OSA roles\n\nThis is a manually generated list of the SHA for the roles.\nThis corresponds to the RC1 to produce.\n\nChange-Id: If34f26d604ff2e52cbbacc31a46f4a8aef6a0258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4eb787a644554b6aabe18b5052594f9fe392aa13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4eb787a644554b6aabe18b5052594f9fe392aa13"}]},"branch":"refs/heads/master"},"e51a11d0548c19bfd1d9f3305a03acaebcaa6463":{"kind":"REWORK","_number":2,"created":"2019-10-21 14:25:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/686581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/686581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/686581/2"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:15:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-21 14:25:04.000000000","tz":180},"subject":"Create train branch for OSA roles","message":"Create train branch for OSA roles\n\nThis is a manually generated list of the SHA for the roles.\nThis corresponds to the RC1 to produce.\n\nChange-Id: If34f26d604ff2e52cbbacc31a46f4a8aef6a0258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e51a11d0548c19bfd1d9f3305a03acaebcaa6463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e51a11d0548c19bfd1d9f3305a03acaebcaa6463"}]},"branch":"refs/heads/master"},"bd3675ce08e33d403f156026d14466a4f6173c94":{"kind":"REWORK","_number":3,"created":"2019-10-21 14:28:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/686581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/686581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/686581/3"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:15:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-21 14:28:25.000000000","tz":180},"subject":"Create train branch for OSA roles","message":"Create train branch for OSA roles\n\nThis is a manually generated list of the SHA for the roles.\nThis corresponds to the RC1 to produce.\n\nChange-Id: If34f26d604ff2e52cbbacc31a46f4a8aef6a0258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd3675ce08e33d403f156026d14466a4f6173c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd3675ce08e33d403f156026d14466a4f6173c94"}]},"branch":"refs/heads/master"},"de0097b91988fcaaa99034cf6a5391b3a5007912":{"kind":"REWORK","_number":4,"created":"2019-10-21 15:21:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/686581/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/686581/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/686581/4"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:15:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-10-21 15:21:27.000000000","tz":0},"subject":"Create train branch for OSA roles","message":"Create train branch for OSA roles\n\nThis is a manually generated list of the SHA for the roles.\nThis corresponds to the RC1 to produce.\n\nChange-Id: If34f26d604ff2e52cbbacc31a46f4a8aef6a0258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de0097b91988fcaaa99034cf6a5391b3a5007912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de0097b91988fcaaa99034cf6a5391b3a5007912"}]},"branch":"refs/heads/master"},"0086a2883d3cb6daf156078067e019acacd6496f":{"kind":"REWORK","_number":5,"created":"2019-10-21 17:48:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/686581/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/686581/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/686581/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/686581/5"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:15:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-21 17:48:05.000000000","tz":180},"subject":"Create train branch for OSA roles","message":"Create train branch for OSA roles\n\nThis is a manually generated list of the SHA for the roles.\nThis corresponds to the RC1 to produce.\n\nChange-Id: If34f26d604ff2e52cbbacc31a46f4a8aef6a0258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0086a2883d3cb6daf156078067e019acacd6496f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0086a2883d3cb6daf156078067e019acacd6496f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
