)]}'
{"id":"openstack%2Freleases~686583","triplet_id":"openstack%2Freleases~master~I89fb0e98b60d422029ebad1fa1dcf936eb6e47e5","project":"openstack/releases","branch":"master","topic":"create-train","hashtags":[],"change_id":"I89fb0e98b60d422029ebad1fa1dcf936eb6e47e5","subject":"Create OSA train","status":"MERGED","created":"2019-10-04 12:23:11.000000000","updated":"2019-10-24 21:33:56.000000000","submitted":"2019-10-24 21:33:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686583-1571952836812-fa70c53b","meta_rev_id":"cbffdc2ce97c2b38d36a7256a1eca89594815adc","_number":686583,"virtual_id_number":686583,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-10-24 09:26:39.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 21:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-24 14:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-10-24 21:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-10-24 21:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-24 21:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 12:23:34.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-23 12:02:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-24 09:26:39.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-24 21:00:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-24 21:24:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-24 21:33:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d14012fefe25e3c5f2c772b4c92fb59d069682d0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:23:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6111d132d6afc1ce56af0aba13fd07ed27fdcae9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:23:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a0060dc6bf702cbf3024595cd5b5bbc0fb596db3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:24:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"92315a76bbc2a817ab8937a976775d791e0e13fc","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-04 12:24:33.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"11c2b4b1cfd497c534010a153da26b90a5352d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 12:34:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a7ac13e49842f89a0ee7e12b330dba : SUCCESS in 8m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/037e2bcbb2a24930a81fbce898ef50df : SUCCESS in 4m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/544008221f78424f949a994a59fc65d3 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3cca30024d3e7194168aed001df33406fd0d4bd0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-21 14:12:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24b6d91ae3682cba90d376496399983342edfc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:32:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7bf62f7b018457d9aa63b9385a4550b : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95022f04374b449d86961d5a05b54579 : FAILURE in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9358002945b8468594ce6b2e9c29f5e1 : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"89f4e8ef910af86e1521d0a62590229e7d81225b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 19:17:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ef7b07b4809dcc7c6c4767cb9a781b34d1de8081","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 19:32:25.000000000","message":"Patch Set 3:\n\nOh, looks like a legit error:\n\nERROR: commit fd131b6cd3b18de5ee9da6adf22f1196cb629927 is not on the master branch but it is listed as the branch point for stable/train to be created","accounts_in_message":[],"_revision_number":3},{"id":"b1745343170008271c71c84efc0e4a07d892580e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:33:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3bf73788b547958cdd03b0bdc5e5c1 : SUCCESS in 8m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68172b674f684c0e87622168c5a55036 : FAILURE in 6m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/30de0b3c21544546ac61db8c5e46bf48 : FAILURE in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b316a30591d9244c5dc8691c93272467460bbb3f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-22 19:46:28.000000000","message":"Patch Set 3:\n\nI\u0027m kinda not understanding whats happening here, as this patch should signify the branch point https://review.opendev.org/#/c/686573.\n\nThat patch also specifies the SHA for all the OSA roles, which i\u0027m not sure are all matching the ones here that have just merged https://review.opendev.org/#/c/686581/ ?","accounts_in_message":[],"_revision_number":3},{"id":"e25815c22df65e758a11ca2d10c82a3543004a0e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 19:52:41.000000000","message":"Patch Set 3:\n\n\u003e I\u0027m kinda not understanding whats happening here, as this patch\n \u003e should signify the branch point https://review.opendev.org/#/c/686573.\n \u003e \n \u003e That patch also specifies the SHA for all the OSA roles, which i\u0027m\n \u003e not sure are all matching the ones here that have just merged\n \u003e https://review.opendev.org/#/c/686581/ ?\n\nThat first patch hasn\u0027t landed yet. Is this an expected commit hash used in the release yaml?","accounts_in_message":[],"_revision_number":3},{"id":"31c63f35b3888d1e1b8ab1e3c3c39bfd0caae89c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-22 19:54:52.000000000","message":"Patch Set 3:\n\n\u003e \u003e I\u0027m kinda not understanding whats happening here, as this patch\n \u003e \u003e should signify the branch point https://review.opendev.org/#/c/686573.\n \u003e \u003e\n \u003e \u003e That patch also specifies the SHA for all the OSA roles, which\n \u003e i\u0027m\n \u003e \u003e not sure are all matching the ones here that have just merged\n \u003e \u003e https://review.opendev.org/#/c/686581/ ?\n \u003e \n \u003e That first patch hasn\u0027t landed yet. Is this an expected commit hash\n \u003e used in the release yaml?\n\nWell this is why i\u0027m confused, as far as i understand we can\u0027t release anything for OSA until https://review.opendev.org/#/c/686573 merges and we have \"frozen\" our roles.","accounts_in_message":[],"_revision_number":3},{"id":"47b6d1c78fdd6188861024e8274c6c59edda8750","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 20:00:29.000000000","message":"Patch Set 3:\n\n\u003e Well this is why i\u0027m confused, as far as i understand we can\u0027t\n \u003e release anything for OSA until https://review.opendev.org/#/c/686573\n \u003e merges and we have \"frozen\" our roles.\n\nMaybe Dmitriy can explain how this needs to work on the OSA side. I unfortunately have no idea what needs to be done where. All I know is the requested release hash here does not appear to be valid.","accounts_in_message":[],"_revision_number":3},{"id":"060b63261b7e46d375bbae9c34beb9f9164b1a6c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-23 12:02:42.000000000","message":"Patch Set 3:\n\nYep, sorry for confusion. This patch waits for https://review.opendev.org/#/c/686573/ and this hash should be placed for RC. I\u0027d set -W, but don\u0027t have such option:(","accounts_in_message":[],"_revision_number":3},{"id":"71e3c8521d131e26091eee40c1f8366f5fe9c304","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-24 07:18:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"bf19b265f8644924a7ef44c58546cca40c24d08a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 07:42:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf0f199b62148a4a1ac15670a777f21 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35bf7704faa04eb5af3924a82250ded1 : SUCCESS in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/db9665131661408990da191cada9bfec : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":4},{"id":"395bb240b35ef890c12a7ce212330902ecc31d52","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-24 09:26:39.000000000","message":"Patch Set 4:\n\nWaiting for evrardjp\u0027s or mnaser\u0027s confirmation","accounts_in_message":[],"_revision_number":4},{"id":"2df423da520843bfc74803f25e9f30854d9a22ae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-24 14:47:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e1d24e228ecebf7fa22c47df00077ad1a5989278","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-24 14:56:17.000000000","message":"Patch Set 4: Code-Review+1\n\nIt seems fair to have another vote before +w this one. It seems okay to me.","accounts_in_message":[],"_revision_number":4},{"id":"a4979e89dc880e696caada7f18d443be092c9722","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-24 21:00:27.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM - the referenced SHA is now merged in OSA and is the correct point to branch from","accounts_in_message":[],"_revision_number":4},{"id":"152d8e3519dfed93ed7fe7d1350f199b155c97e3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-24 21:24:05.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3a352631ffd902dcb9f1709217b792180d243276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:24:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"876cc40eba4dc70a3821d6a68aeb67fa61f0737d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-24 21:33:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3433fd5faf3e9371aa4e4cfaab6ff23692a47159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:33:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"24cb5e0016065cfb7f46f1831702171d79b1373b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:33:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c32a23081c44d18caa07e5df6eadcd : SUCCESS in 8m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a9d2139b2bf456bab59900f407bd922 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a19891b8b7522df87424d209dd613875e402b6df","revisions":{"ace65163cfbdd6081bd60af6f24c2945deb9983c":{"kind":"REWORK","_number":1,"created":"2019-10-04 12:23:11.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/83/686583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/686583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/686583/1"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:18:12.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:21:04.000000000","tz":120},"subject":"[DNM] Example patch -- Create OSA train","message":"[DNM] Example patch -- Create OSA train\n\nOSA is lagging behind.\nWhat needs to happen:\n- OSA should first have its milestone merged\n- OSA should propose a m1 release\n- OSA should merge https://review.opendev.org/#/c/686573/1\n- OSA should edit this patch to the appropriate sha on top of\n  https://review.opendev.org/#/c/686573/1\n\nChange-Id: I89fb0e98b60d422029ebad1fa1dcf936eb6e47e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ace65163cfbdd6081bd60af6f24c2945deb9983c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ace65163cfbdd6081bd60af6f24c2945deb9983c"}]},"branch":"refs/heads/master"},"fd131b6cd3b18de5ee9da6adf22f1196cb629927":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-04 12:24:30.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/83/686583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/686583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/686583/2"}}},"commit":{"parents":[{"commit":"79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a","subject":"Merge \"improve readability of schedule table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b8f22b7eb435b328dcf5df1d72b3eb34d5d92a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:18:12.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:24:28.000000000","tz":0},"subject":"[DNM] Example patch -- Create OSA train","message":"[DNM] Example patch -- Create OSA train\n\nOSA is lagging behind.\nWhat needs to happen:\n- We should first have its milestone merged\n- We should propose a m1 release\n- We should merge https://review.opendev.org/#/c/686573/1\n- We should edit this patch to the appropriate sha on top of\n  https://review.opendev.org/#/c/686573/1\n- We should branch the roles with https://review.opendev.org/#/c/686581/\n\nChange-Id: I89fb0e98b60d422029ebad1fa1dcf936eb6e47e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd131b6cd3b18de5ee9da6adf22f1196cb629927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd131b6cd3b18de5ee9da6adf22f1196cb629927"}]},"branch":"refs/heads/master"},"5dd0d1205ca90e17e48cc95b2274325d9b2c4565":{"kind":"REWORK","_number":3,"created":"2019-10-21 14:12:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/686583/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/686583/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/686583/3"}}},"commit":{"parents":[{"commit":"e4adf7b08e298c716bb46abdf6439a62996fc302","subject":"Merge \"TripleO RC tags + stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4adf7b08e298c716bb46abdf6439a62996fc302"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:18:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-21 14:02:32.000000000","tz":180},"subject":"Create OSA train","message":"Create OSA train\n\nOSA is lagging behind.\n\nWhat needs to happen:\n- We should branch the roles with https://review.opendev.org/#/c/686581/\n\nChange-Id: I89fb0e98b60d422029ebad1fa1dcf936eb6e47e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5dd0d1205ca90e17e48cc95b2274325d9b2c4565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5dd0d1205ca90e17e48cc95b2274325d9b2c4565"}]},"branch":"refs/heads/master"},"a19891b8b7522df87424d209dd613875e402b6df":{"kind":"REWORK","_number":4,"created":"2019-10-24 07:18:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/686583/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/686583/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/686583/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/686583/4"}}},"commit":{"parents":[{"commit":"e4adf7b08e298c716bb46abdf6439a62996fc302","subject":"Merge \"TripleO RC tags + stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4adf7b08e298c716bb46abdf6439a62996fc302"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-10-04 12:18:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-10-24 07:18:22.000000000","tz":0},"subject":"Create OSA train","message":"Create OSA train\n\nOSA is lagging behind.\n\nWhat needs to happen:\n- We should branch the roles with https://review.opendev.org/#/c/686581/\n\nChange-Id: I89fb0e98b60d422029ebad1fa1dcf936eb6e47e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a19891b8b7522df87424d209dd613875e402b6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a19891b8b7522df87424d209dd613875e402b6df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
