)]}'
{"id":"openstack%2Freleases~687097","triplet_id":"openstack%2Freleases~master~I39f64af8eae7aacae0bfdc9196668d41e4db8312","project":"openstack/releases","branch":"master","topic":"reno","hashtags":[],"change_id":"I39f64af8eae7aacae0bfdc9196668d41e4db8312","subject":"Add recent release note landing pages","status":"MERGED","created":"2019-10-07 16:49:22.000000000","updated":"2019-10-09 14:53:38.000000000","submitted":"2019-10-09 14:53:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687097-1570632818811-25c6a380","meta_rev_id":"1eca5fb451e0d9b0bd5f0eaf9f56488c3f823a88","_number":687097,"virtual_id_number":687097,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-08 14:40:31.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 14:53:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-09 14:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-10-08 14:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-09 14:39:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 14:37:11.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-10-09 14:39:30.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-09 14:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfb02bfe8d92b1390af4f498f3ccd29ff2c20275","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-07 16:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9c2654b2602b80339bfdfd782bb5909ffd4c982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 17:03:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb08cb4c6914a9faf15ce105e98c42f : SUCCESS in 8m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ae41b19365145708d16651652addd47 : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6c93ba9348d94461aa8eff6136f70501 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"35ecde26a86bffbc55578ab74f38a9206c5a8d8d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-08 14:37:11.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks okay to me.","accounts_in_message":[],"_revision_number":1},{"id":"f5c3c5dd4bc679d2ad66cbf6e356dbd829d75b16","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-10-08 14:38:09.000000000","message":"Patch Set 1:\n\nI didn\u0027t check that the links are valid themselves, one per one. I assume the projects have merged the appropriate patches.\n\nMaybe we should add validation of those links if it\u0027s not done (not sure if it\u0027s done)...","accounts_in_message":[],"_revision_number":1},{"id":"c8a0bdc670ed6069ae2af4e878c4b04060a3a3b7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-08 14:39:22.000000000","message":"Patch Set 1:\n\nThese are added by our tool/script that goes through each one and validates that the landing pages are there, so should be safe to assume it\u0027s all good.\n\nI think we actually do have some validation in the job too that will error if it gets a 404 on anything.","accounts_in_message":[],"_revision_number":1},{"id":"998141722c73ab4c958e1ddfecf5d44383c561c4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-08 14:40:31.000000000","message":"Patch Set 1:\n\nhttps://opendev.org/openstack/releases/src/branch/master/openstack_releases/cmds/validate.py#L513","accounts_in_message":[],"_revision_number":1},{"id":"9ea65f6493a83130b5d798ff9332e0c54991997a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-09 14:39:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7d39b5cb01e18dcb6a8a477c321ba6d3b4aa7fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca2155d5fbd86ae4ab8125a200e141f994c4caa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:53:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f83fb23ccf5ccd189868ff784affc78717bc6149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:53:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a18fe82cc14014bfbce5f8d163e59a : SUCCESS in 8m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8e07694614948b5833f59391b8ba8ab : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"984c2f6f5c65d8a2acc44b238819a132c714d193","revisions":{"984c2f6f5c65d8a2acc44b238819a132c714d193":{"kind":"REWORK","_number":1,"created":"2019-10-07 16:49:22.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/97/687097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/687097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/687097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/687097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/687097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/687097/1"}}},"commit":{"parents":[{"commit":"b2afe9e06f4780b67972f7a0c286060312d99cc1","subject":"Merge \"Update designate liaison email\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b2afe9e06f4780b67972f7a0c286060312d99cc1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-07 16:49:17.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-07 16:49:17.000000000","tz":-300},"subject":"Add recent release note landing pages","message":"Add recent release note landing pages\n\nChange-Id: I39f64af8eae7aacae0bfdc9196668d41e4db8312\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/984c2f6f5c65d8a2acc44b238819a132c714d193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/984c2f6f5c65d8a2acc44b238819a132c714d193"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
