)]}'
{"id":"openstack%2Freleases~687337","triplet_id":"openstack%2Freleases~master~I72ddf92837bfd99beb4279afabf57bb84c7792fe","project":"openstack/releases","branch":"master","topic":"puppet-openstack","hashtags":[],"change_id":"I72ddf92837bfd99beb4279afabf57bb84c7792fe","subject":"Release puppet-designate 15.5.0 for Train","status":"MERGED","created":"2019-10-08 15:39:42.000000000","updated":"2019-10-10 17:15:25.000000000","submitted":"2019-10-10 17:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687337-1570727725324-9e252c24","meta_rev_id":"eeff2346a4d6668c9e99e1293b75a4e0fa0cd0e2","_number":687337,"virtual_id_number":687337,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2019-10-10 11:39:35.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 17:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-10 13:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-10 17:01:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 09:37:53.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-09 13:29:51.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-10-10 17:01:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-10 17:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c85669c6c7bd7ef5a0e5dfc5e866e3b8bda02e01","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:39:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"307aa2b2bf08e234a8ae77ebffc2b4aa33cbd740","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-08 15:40:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"317a40d32bd123aceb06cf078207766d11fd087a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:58:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8296b038abe248d3a9fd883fdad4e8c3 : SUCCESS in 11m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8fb99d31f674bf2b8e612b6332fa6f2 : FAILURE in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aeca46b04ccb4f1994aa41956921e837 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3f40d155f9b2194f994106683bdf3ecd3e142cc2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-09 09:38:00.000000000","message":"Patch Set 2:\n\nThoughts/ack on using 15.5.0?","accounts_in_message":[],"_revision_number":2},{"id":"48256f810f1658234afed56a29aec42ba14f9f32","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-09 19:17:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"01325b59218729e2463528612efb87e625d198ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 19:29:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65c130e57664bb08e71755ff68763b1 : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9bd4906f9ef84746983bcfc646ea14cd : FAILURE in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3b6760c9525e49ba9dc767e80cfd9e18 : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3bf77a03b46780dccf8b3ec2db1d5d71ed5e3e96","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 19:41:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e36c25ec06b16d31431812a355e2ec014d51f62","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 11:39:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"edccc7ee5ef5d8a66a650bb4f00c1e9a86a4c63a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-10 11:39:35.000000000","message":"Patch Set 3:\n\nThanks Sean!","accounts_in_message":[],"_revision_number":3},{"id":"1ea6f979e2a5ab774757e5c34bebff3618c9b317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 11:52:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5db408d2d94d53a378bf7811f1853e : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49646deaadbc45edaef56ebe836259b8 : SUCCESS in 4m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/be931fc2e50540889ad5004799997585 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9213021b9b81947447c0d7a53ad01bea80dac368","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 13:41:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"58e5010fb6d12e8e9f53c21d2ed41f6b478fae05","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 17:01:54.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"804b85d6dc8cacefdb692c14d20f3d8b39ff1ee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 17:02:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3e292d4cc4e420e3d9c5a56ced5874ca085cde9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 17:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"86499c48403b9e5d4f00c52b115d2a85af2ae778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 17:15:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae183c3cb4c14329baca15ed900dd41c : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be6072e20bb347d694372576b1acdebe : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1bccfc6df87df12a0ea0a29f374261d054da71c5","revisions":{"14e8181fe5989b4ec07f1ba6898c17d046a98a7f":{"kind":"REWORK","_number":1,"created":"2019-10-08 15:39:42.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/37/687337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/687337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/687337/1"}}},"commit":{"parents":[{"commit":"0f2937e9c29a1d112bcc957436485a09b87a622f","subject":"Merge \"Add Ussuri cycle schedule for manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f2937e9c29a1d112bcc957436485a09b87a622f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:36:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:36:59.000000000","tz":120},"subject":"Release puppet-designate 15.5.0 for Train","message":"Release puppet-designate 15.5.0 for Train\n\nWe need to release 15.5.0 instead of 15.4.0\nfor Train because we had broken code in that\nrelease since Designate has removed the\npool-manager [1].\n\nWe fixed our testing in [2] and deprecated\nthat code and added a release note in [3].\n\nSee the release patch that updates the\nmetadata [4] which holds the complete\npatch chain.\n\n[1] https://docs.openstack.org/releasenotes/designate/train.html\n[2] https://review.opendev.org/#/c/687247/\n[3] https://review.opendev.org/#/c/687333/\n[4] https://review.opendev.org/#/c/687334\n\nChange-Id: I72ddf92837bfd99beb4279afabf57bb84c7792fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14e8181fe5989b4ec07f1ba6898c17d046a98a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14e8181fe5989b4ec07f1ba6898c17d046a98a7f"}]},"branch":"refs/heads/master"},"92c8348789d41ecfec965be49995b900f2146f14":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-08 15:40:14.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/37/687337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/687337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/687337/2"}}},"commit":{"parents":[{"commit":"0f2937e9c29a1d112bcc957436485a09b87a622f","subject":"Merge \"Add Ussuri cycle schedule for manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f2937e9c29a1d112bcc957436485a09b87a622f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:36:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:40:03.000000000","tz":120},"subject":"Release puppet-designate 15.5.0 for Train","message":"Release puppet-designate 15.5.0 for Train\n\nWe need to release 15.5.0 instead of 15.4.0\nfor Train because we had broken code in that\nrelease since Designate has removed the\npool-manager [1].\n\nWe fixed our testing in [2] and deprecated\nthat code and added a release note in [3].\n\nSee the release patch that updates the\nmetadata [4] which holds the complete\npatch chain.\n\n[1] https://docs.openstack.org/releasenotes/designate/train.html\n[2] https://review.opendev.org/#/c/687247/\n[3] https://review.opendev.org/#/c/687333/\n[4] https://review.opendev.org/#/c/687334\n\nDepends-On: https://review.opendev.org/#/c/687334\nChange-Id: I72ddf92837bfd99beb4279afabf57bb84c7792fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92c8348789d41ecfec965be49995b900f2146f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92c8348789d41ecfec965be49995b900f2146f14"}]},"branch":"refs/heads/master"},"1bccfc6df87df12a0ea0a29f374261d054da71c5":{"kind":"REWORK","_number":3,"created":"2019-10-10 11:39:26.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/37/687337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/687337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/687337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/687337/3"}}},"commit":{"parents":[{"commit":"0f2937e9c29a1d112bcc957436485a09b87a622f","subject":"Merge \"Add Ussuri cycle schedule for manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f2937e9c29a1d112bcc957436485a09b87a622f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-08 15:36:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-10-10 11:39:24.000000000","tz":0},"subject":"Release puppet-designate 15.5.0 for Train","message":"Release puppet-designate 15.5.0 for Train\n\nWe need to release 15.5.0 instead of 15.4.0\nfor Train because we had broken code in that\nrelease since Designate has removed the\npool-manager [1].\n\nWe fixed our testing in [2] and deprecated\nthat code and added a release note in [3].\n\nSee the release patch that updates the\nmetadata [4] which holds the complete\npatch chain.\n\n[1] https://docs.openstack.org/releasenotes/designate/train.html\n[2] https://review.opendev.org/#/c/687247/\n[3] https://review.opendev.org/#/c/687333/\n[4] https://review.opendev.org/#/c/687334\n\nDepends-On: https://review.opendev.org/#/c/687334\nChange-Id: I72ddf92837bfd99beb4279afabf57bb84c7792fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bccfc6df87df12a0ea0a29f374261d054da71c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1bccfc6df87df12a0ea0a29f374261d054da71c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
