)]}'
{"id":"openstack%2Freleases~687576","triplet_id":"openstack%2Freleases~master~I532ad8bf7d459cd3de36c617950caa0cb971df2a","project":"openstack/releases","branch":"master","topic":"train-rc2","hashtags":[],"change_id":"I532ad8bf7d459cd3de36c617950caa0cb971df2a","subject":"Release heat train rc2","status":"MERGED","created":"2019-10-09 14:17:26.000000000","updated":"2019-10-09 20:02:28.000000000","submitted":"2019-10-09 20:02:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687576-1570651348605-25c6a380","meta_rev_id":"071b873f3c8f550fc734c3ebfd69bdcb706c7fdc","_number":687576,"virtual_id_number":687576,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 20:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-09 14:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-09 16:28:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-09 19:49:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 16:28:09.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-10-09 19:49:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-09 20:02:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"204ce61db5d550c7150f99e6750adc24a1dad71d","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-10-09 14:17:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdaf8e8eef129cc392403d0186f3950145f2754e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:29:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e8d7e44c04431ba6b5e1d547a4c536 : FAILURE in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a7c4943e3bd4f7d99ae56799d0bb5ed : FAILURE in 3m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9a00b7bc16da4d0691410f59ed1a26f8 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"81e3470aad807a33d482e1a165e6088dc83cc25d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 14:40:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8c2913c5da20b83600d3d2c150979d5f431a48db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 14:54:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2884c3b9d3424dbcb6bdc8fa60370b10 : SUCCESS in 9m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d100e362dd548e3a9e126ab04149193 : SUCCESS in 4m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/69e1932c86864bd7aab4129559b864da : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"92c38d37d943913d3e70332df430e9b17e8edaa3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 14:57:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"740728209d70965dff73fb63d8cc43437be11294","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 16:11:48.000000000","message":"Topic changed from release-heat to train-rc2","accounts_in_message":[],"_revision_number":2},{"id":"9625f30f3e36c64cd0e94245aa30f51da751cdf4","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-09 16:28:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6e54f279d30b393eb4f4d7b757ceb3c4434a4189","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 19:49:27.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"119284af03ef663066d043857540449ff1f571fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 19:49:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"385eca77d2c8f4c3373d4be621b56626f7f232fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 20:02:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1535ca97c68470e8cb1a0e668e2a30b : SUCCESS in 8m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1b0e4cace674aff9cfd0db50de7047e : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"071b873f3c8f550fc734c3ebfd69bdcb706c7fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 20:02:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0e66121c04dc9adc49bce329820b99e65521146","revisions":{"d08cca805d36d1c105876ba0ae46a221eb3dd983":{"kind":"REWORK","_number":1,"created":"2019-10-09 14:17:26.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/76/687576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/687576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/687576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/687576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/687576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/687576/1"}}},"commit":{"parents":[{"commit":"eb87a7ab3522ddd75c36eb541522d2c5f25f400a","subject":"Merge \"Branch rpm-packaging for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb87a7ab3522ddd75c36eb541522d2c5f25f400a"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-10-09 14:16:58.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-10-09 14:16:58.000000000","tz":480},"subject":"Release heat train rc2","message":"Release heat train rc2\n\nChange-Id: I532ad8bf7d459cd3de36c617950caa0cb971df2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08cca805d36d1c105876ba0ae46a221eb3dd983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d08cca805d36d1c105876ba0ae46a221eb3dd983"}]},"branch":"refs/heads/master"},"d0e66121c04dc9adc49bce329820b99e65521146":{"kind":"REWORK","_number":2,"created":"2019-10-09 14:40:28.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/76/687576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/687576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/687576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/687576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/687576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/687576/2"}}},"commit":{"parents":[{"commit":"eb87a7ab3522ddd75c36eb541522d2c5f25f400a","subject":"Merge \"Branch rpm-packaging for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb87a7ab3522ddd75c36eb541522d2c5f25f400a"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2019-10-09 14:16:58.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 14:40:24.000000000","tz":0},"subject":"Release heat train rc2","message":"Release heat train rc2\n\nChange-Id: I532ad8bf7d459cd3de36c617950caa0cb971df2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0e66121c04dc9adc49bce329820b99e65521146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0e66121c04dc9adc49bce329820b99e65521146"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
