)]}'
{"id":"openstack%2Freleases~687590","triplet_id":"openstack%2Freleases~master~Ica81d1eb8b9e566b98ea60bf7d35c81e46611760","project":"openstack/releases","branch":"master","topic":"train-rc2","hashtags":[],"change_id":"Ica81d1eb8b9e566b98ea60bf7d35c81e46611760","subject":"Release mistral RC2 for Train","status":"MERGED","created":"2019-10-09 15:24:24.000000000","updated":"2019-10-10 14:07:57.000000000","submitted":"2019-10-10 14:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687590-1570716477101-fe29816b","meta_rev_id":"d4cf5a35dcf8bb50c33a90efcb9c3fbb96a92b2d","_number":687590,"virtual_id_number":687590,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 14:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-10 13:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-10 08:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-10-10 07:28:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-10 13:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 07:28:52.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-10 08:49:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-10 14:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"788170df491b7bef6c62ad97935617618e0784a9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 15:24:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbad50c09303e62e7028e6a2ebd210a3a2edebb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:43:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c5fe03a76a48048e8bc0c604ce0d1d : SUCCESS in 9m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/643b8b23fa9c420f93b140d3f2f71f67 : SUCCESS in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8cc3357a60f247b4946618c7bb4bad3b : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"21b30b627b1a6b6036287cca74c47f1ee7901b02","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-10 07:28:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f3c37dddd734afa1593f6e6fa5e80fca0ce0a643","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-10-10 07:28:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"42c9113e5625d32567da054521db637f3cec5610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 07:38:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fed3679c1604441ab91ab0005e0506d : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52a1183df0d24cfcbe462b71172e9028 : SUCCESS in 8m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7a3659f62bb84810bfeba5e1c57dfc48 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0902d65763dd0376909fab1ebde6f300ac04ef64","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-10 08:49:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81690e92527ab51e2943c5d4302a2db446cef686","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 13:57:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nAll good, but just pointing out that the gitreview and tox updates are non-functional changes that you wouldn\u0027t need to wait to pick up for final release. Not saying you did or that you shouldn\u0027t have updated, just making sure you don\u0027t think you have to quick rush those in for the future. ;)","accounts_in_message":[],"_revision_number":2},{"id":"86e2b1533c573d624336eba526d83f28d089dc88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:58:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85061e596bc8ea19acb3832c4a8ccd40f363c79d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:07:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58973bc12e1a4433a5a1e2160c3ce340 : SUCCESS in 8m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af4a218d0a634cde8159051be84fb379 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b18984c8dde5efaa387410f008b30b9c8e42c934","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:07:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb345fb7f7aea118927f76103f37465630651aee","revisions":{"bf147cc1b1255bb039688e44734aa79ee20fa2bd":{"kind":"REWORK","_number":1,"created":"2019-10-09 15:24:24.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/90/687590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/687590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/687590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/687590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/687590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/687590/1"}}},"commit":{"parents":[{"commit":"7272fc3720fb584b30dd2558977995aecdf7f8d7","subject":"Merge \"nova: release train 20.0.0.0rc2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7272fc3720fb584b30dd2558977995aecdf7f8d7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:24:07.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:24:11.000000000","tz":-300},"subject":"Release mistral RC2 for Train","message":"Release mistral RC2 for Train\n\nThis release is being proposed to pick up translations and bugfixes\nthat have merged to stable/train but not been released.\n\nNot all changes are critical to make it into the initial Train\ncoordinated release and can wait until a stable release later. If\nthis patch does not get an ack from the PTL and/or release liaison\nby EOD Thursday it will just be abandoned.\n\nTeams may also choose to -1 this patch and update it once any final\ncommits merge prior to the October 10 deadline.\n\nChange-Id: Ica81d1eb8b9e566b98ea60bf7d35c81e46611760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf147cc1b1255bb039688e44734aa79ee20fa2bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf147cc1b1255bb039688e44734aa79ee20fa2bd"}]},"branch":"refs/heads/master"},"cb345fb7f7aea118927f76103f37465630651aee":{"kind":"REWORK","_number":2,"created":"2019-10-10 07:28:45.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/90/687590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/687590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/687590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/687590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/687590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/687590/2"}}},"commit":{"parents":[{"commit":"7272fc3720fb584b30dd2558977995aecdf7f8d7","subject":"Merge \"nova: release train 20.0.0.0rc2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7272fc3720fb584b30dd2558977995aecdf7f8d7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:24:07.000000000","tz":-300},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-10-10 07:28:17.000000000","tz":0},"subject":"Release mistral RC2 for Train","message":"Release mistral RC2 for Train\n\nThis release is being proposed to pick up translations and bugfixes\nthat have merged to stable/train but not been released.\n\nNot all changes are critical to make it into the initial Train\ncoordinated release and can wait until a stable release later. If\nthis patch does not get an ack from the PTL and/or release liaison\nby EOD Thursday it will just be abandoned.\n\nTeams may also choose to -1 this patch and update it once any final\ncommits merge prior to the October 10 deadline.\n\nChange-Id: Ica81d1eb8b9e566b98ea60bf7d35c81e46611760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb345fb7f7aea118927f76103f37465630651aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb345fb7f7aea118927f76103f37465630651aee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
