)]}'
{"id":"openstack%2Freleases~687596","triplet_id":"openstack%2Freleases~master~I572379e09dd980abf475b3191cd0ce415cc6668d","project":"openstack/releases","branch":"master","topic":"train-rc2","hashtags":[],"change_id":"I572379e09dd980abf475b3191cd0ce415cc6668d","subject":"Release sahara and sahara-plugin-* RC2 for Train","status":"MERGED","created":"2019-10-09 15:37:42.000000000","updated":"2019-10-10 20:24:19.000000000","submitted":"2019-10-10 20:24:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687596-1570739059698-14372de2","meta_rev_id":"20ca6b456975d54d00a001440e35ebde16447f86","_number":687596,"virtual_id_number":687596,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"date":"2019-10-10 20:00:05.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 20:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-10 20:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":1,"date":"2019-10-10 20:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-10 20:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 15:52:46.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-10-10 20:00:05.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-10-10 20:00:46.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-10-10 20:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0720df14180d88dd89330f70a9606829072e38ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 15:37:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ac23f68ab2acafd28ccef3d3bd917dff5b5f9cd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-09 15:45:03.000000000","message":"Patch Set 1:\n\nUhm, Jeremy, can you please manage to ping Tony later, so that we can merge https://review.opendev.org/#/c/686732/ and include it here? (and maybe also the sahara.git change)","accounts_in_message":[],"_revision_number":1},{"id":"63747d75d3ac1bbbc4d2cf9afabe399c104d4e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:49:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45eb5caf8ec434a8ac9d61159e45e4b : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea1108234a674224b897b8c39cc1d851 : SUCCESS in 9m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/87ee11f1d50f448e8e70e9b0c314c67d : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a861f57c8f672c6b2cebfea45a0b481740e77df4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 15:55:26.000000000","message":"Patch Set 1: Code-Review-1\n\nWaiting for final patches to land. Please update this with the new commit hash when ready. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"02f625dcf1c0bf5e090bb2f4b6f019557442b403","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 14:18:05.000000000","message":"Patch Set 1:\n\nLooks like referenced patch landed. Are you ready to update the commit hash and get this through? Or are there other important patches in flight?","accounts_in_message":[],"_revision_number":1},{"id":"1951fd2a24417b9f8851218ed6f8259568ba070a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-10 15:26:07.000000000","message":"Patch Set 1:\n\nJeremy, I\u0027m going to amend the patch to include the cdh change. If you want to merge the grenade patches, I think that they are clear enough that we can get just one +2 (also because Telles is not around). If you manage to merge it, please change this review to include that patch as well (or propose a new patch - I will be offline for a few hours).\n\nSean, thanks for the patience!","accounts_in_message":[],"_revision_number":1},{"id":"4a89bc2030ec8009241dc28528412a0b107b1520","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-10 15:29:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd157c93eaf9aea36a05ff19242a480ba08fa6c3","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-10-10 15:38:45.000000000","message":"Patch Set 2: Code-Review+1\n\nthis can be merged.\n\ni will propose another patch here after https://review.opendev.org/#/c/687163/ merges (dependent on https://review.opendev.org/#/c/687696/ and https://review.opendev.org/#/c/687693/ )","accounts_in_message":[],"_revision_number":2},{"id":"3f2f3c40b65d82b23a01c4068966c0c0a69bf575","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 15:39:58.000000000","message":"Patch Set 2:\n\n\u003e this can be merged.\n \u003e \n \u003e i will propose another patch here after https://review.opendev.org/#/c/687163/\n \u003e merges (dependent on https://review.opendev.org/#/c/687696/ and\n \u003e https://review.opendev.org/#/c/687693/ )\n\nIf you would be looking at an RC3 release yet today, we might as well just wait. We don\u0027t want to do RCs if no one is going to get a chance to pick them up.","accounts_in_message":[],"_revision_number":2},{"id":"7da9e36607d402acf6f4b17bd21dbc39ad129a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:41:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf3e65a66374e21a7596eeffcfd4e6a : SUCCESS in 9m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5027b0719d947a8ad9799745b5f7c07 : SUCCESS in 7m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/41e6d2a845d04d6a917db85d0d420eb7 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ed33dc8f79c820a25246efb9a47fbb4b7d2679c2","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-10-10 15:45:00.000000000","message":"Patch Set 2: -Code-Review\n\nok, let\u0027s wait (just a couple hours), good point","accounts_in_message":[],"_revision_number":2},{"id":"f3d6f7556cb03883881b4d7a0f1aadd8d20225c2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 15:46:33.000000000","message":"Patch Set 2:\n\n\u003e ok, let\u0027s wait (just a couple hours), good point\n\nWe\u0027ll hold the train for you. ;)","accounts_in_message":[],"_revision_number":2},{"id":"5e30148ccbb65c97cb2af26a1c56f6a06046b52d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 19:44:45.000000000","message":"Patch Set 2:\n\nLooks like the last patch is just waiting on a non-voting patch to land. Anything needed with these plugins after that, or was it just a matter of making sure sahara changes didn\u0027t break anything with the plugins?","accounts_in_message":[],"_revision_number":2},{"id":"96f27856b4bc97fde7deb141fb58b107e4265272","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-10 19:59:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e02c6d615d5dccf46b2f9c2a22e2f987b4ad015","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-10 19:59:56.000000000","message":"Patch Set 3:\n\nThat should be it :) now we are more python3-happy.","accounts_in_message":[],"_revision_number":3},{"id":"ffcab1fef1515cfaed09020175255c49f6e4b188","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-10 20:00:05.000000000","message":"Patch Set 3:\n\nand thanks again for the patience","accounts_in_message":[],"_revision_number":3},{"id":"4c9376971556e47e0e37029f29f9957b76de5232","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-10-10 20:00:46.000000000","message":"Patch Set 3: Code-Review+1\n\nwow... luigi is faster than me... had literally typed \"git revie\" when i got the notification\nto answer Sean\u0027s question: no, we don\u0027t need any rc3. just that sahara patch which luigi just put. this is good to go.","accounts_in_message":[],"_revision_number":3},{"id":"a8dc319b196d66d4519d251001704398364233ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:12:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e920ed318a41d685de5c645758b9f5 : SUCCESS in 8m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cee5d6746bb44d37af2ae70c64c8b253 : SUCCESS in 6m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/884cbde419714f178e7ab9c1a656a4a0 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f8ac972a8ec6c21fb4d1bbb16388a1cf56629032","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 20:16:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nOK, I was just confused. Looks like we could have processed the plugin releases earlier as sahara could just be done by itself. But now we have them all, so good that it\u0027s all covered.","accounts_in_message":[],"_revision_number":3},{"id":"efaa64caf337b46606717f3a5fbcf227cfdc7734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:16:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01418f6d371162d9c6a8497fa577d2eec5115df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:24:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402c85f1d6204cd2819df9dc49ec9660 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af5960b1862645d7b56d4f606bfd8857 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c2148e272b424debb0448962d36a624975d71423","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:24:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"945a6bda32724783ed644b725a8dfee96439206b","revisions":{"281cfc2802abc7c71d2df50fea1080ed4d715893":{"kind":"REWORK","_number":1,"created":"2019-10-09 15:37:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/96/687596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/687596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/687596/1"}}},"commit":{"parents":[{"commit":"7272fc3720fb584b30dd2558977995aecdf7f8d7","subject":"Merge \"nova: release train 20.0.0.0rc2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7272fc3720fb584b30dd2558977995aecdf7f8d7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:37:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:37:28.000000000","tz":-300},"subject":"Release sahara-plugin-* RC2 for Train","message":"Release sahara-plugin-* RC2 for Train\n\nThis release is being proposed to pick up translations and bugfixes\nthat have merged to stable/train but not been released.\n\nNot all changes are critical to make it into the initial Train\ncoordinated release and can wait until a stable release later. If\nthis patch does not get an ack from the PTL and/or release liaison\nby EOD Thursday it will just be abandoned.\n\nTeams may also choose to -1 this patch and update it once any final\ncommits merge prior to the October 10 deadline.\n\nChange-Id: I572379e09dd980abf475b3191cd0ce415cc6668d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/281cfc2802abc7c71d2df50fea1080ed4d715893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/281cfc2802abc7c71d2df50fea1080ed4d715893"}]},"branch":"refs/heads/master"},"0b17a935d4c09454e9d5b30a1f5594767bcb748a":{"kind":"REWORK","_number":2,"created":"2019-10-10 15:29:08.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/96/687596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/687596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/687596/2"}}},"commit":{"parents":[{"commit":"7272fc3720fb584b30dd2558977995aecdf7f8d7","subject":"Merge \"nova: release train 20.0.0.0rc2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7272fc3720fb584b30dd2558977995aecdf7f8d7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:37:22.000000000","tz":-300},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-10 15:28:34.000000000","tz":120},"subject":"Release sahara-plugin-* RC2 for Train","message":"Release sahara-plugin-* RC2 for Train\n\nThis release is being proposed to pick up translations and bugfixes\nthat have merged to stable/train but not been released.\n\nNot all changes are critical to make it into the initial Train\ncoordinated release and can wait until a stable release later. If\nthis patch does not get an ack from the PTL and/or release liaison\nby EOD Thursday it will just be abandoned.\n\nTeams may also choose to -1 this patch and update it once any final\ncommits merge prior to the October 10 deadline.\n\nChange-Id: I572379e09dd980abf475b3191cd0ce415cc6668d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b17a935d4c09454e9d5b30a1f5594767bcb748a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b17a935d4c09454e9d5b30a1f5594767bcb748a"}]},"branch":"refs/heads/master"},"945a6bda32724783ed644b725a8dfee96439206b":{"kind":"REWORK","_number":3,"created":"2019-10-10 19:59:23.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/96/687596/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/687596/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/687596/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/687596/3"}}},"commit":{"parents":[{"commit":"7272fc3720fb584b30dd2558977995aecdf7f8d7","subject":"Merge \"nova: release train 20.0.0.0rc2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7272fc3720fb584b30dd2558977995aecdf7f8d7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-09 15:37:22.000000000","tz":-300},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-10 19:58:49.000000000","tz":120},"subject":"Release sahara and sahara-plugin-* RC2 for Train","message":"Release sahara and sahara-plugin-* RC2 for Train\n\nThis release is being proposed to pick up translations and bugfixes\nthat have merged to stable/train but not been released.\n\nNot all changes are critical to make it into the initial Train\ncoordinated release and can wait until a stable release later. If\nthis patch does not get an ack from the PTL and/or release liaison\nby EOD Thursday it will just be abandoned.\n\nTeams may also choose to -1 this patch and update it once any final\ncommits merge prior to the October 10 deadline.\n\nChange-Id: I572379e09dd980abf475b3191cd0ce415cc6668d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/945a6bda32724783ed644b725a8dfee96439206b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/945a6bda32724783ed644b725a8dfee96439206b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
