)]}'
{"id":"openstack%2Freleases~687992","triplet_id":"openstack%2Freleases~master~I93747459c5e4555ea3e97a02bb49a2f3e400d6d1","project":"openstack/releases","branch":"master","topic":"train-final","hashtags":[],"change_id":"I93747459c5e4555ea3e97a02bb49a2f3e400d6d1","subject":"Mark Train as released","status":"MERGED","created":"2019-10-10 22:06:36.000000000","updated":"2019-10-16 14:00:30.000000000","submitted":"2019-10-16 14:00:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687992-1571234430753-29eb8155","meta_rev_id":"1b5659a5aaf4247f64d16dc9ce08af7ace8f20f7","_number":687992,"virtual_id_number":687992,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 14:00:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-16 12:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-16 13:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-10-16 13:45:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-16 13:45:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-16 13:45:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 07:15:27.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-10-16 13:45:23.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-16 13:45:23.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-10-16 13:45:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-10-16 14:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6403a9e68ad90e694594d9e99724a13add051f61","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-10 22:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73b24d56f064d2192e2dc31a2286d45549f4422f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 22:19:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d69916c59c44485840e07512a76e1e8 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/903c55c0573e4ed6929ae925b034f26d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6cef8a44c8421d94276ffadbe4227d : SUCCESS in 8m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f361637d373141e981f0be3a4a1f48d9 : SUCCESS in 4m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7b3393be91b4442fb718b527b6296cba : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1961de9bed84a6eb9429a8d91c55484 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9f2568fe011c03c81551e074f6e100c318b037f3","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-15 07:15:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"01fd3de61b76699a00f2118696969e27a27c2237","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-15 07:24:39.000000000","message":"Topic novaclient-ussuri-16.0.0 removed","accounts_in_message":[],"_revision_number":2},{"id":"80dc26d94dd846efb6147d93e2c0cdbc60f0cff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 07:38:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce1dc425d8674828bb446cf7ec060133 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bce39de4da24f22a0338f28d949a2cf : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebe87a965794b5ca21b68ca0353316a : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f5c8ccff7b247ce8f8f2fc1a85ce972 : SUCCESS in 4m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/58ebc2aae36445cfb24afbacc09e290f : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e796d743acfc42bd895c42a277c8f46e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"690edd79324b65e31b5e4514b4c83cc42fd9797b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-15 13:43:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"302bc8102412362b17dff093b59d42175dc5469b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 13:56:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56c9dff877914596a883cd62c7ddcfa3 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0d4a01f59b54b37980d9698edace792 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224ff8fd2e6b469eb752f8dec345888e : SUCCESS in 10m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6544de8fd5d43caa084d6f0a5e67432 : SUCCESS in 7m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dc9127c983654c28954a0ef6ac8c71bc : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0207b1bc1c6e43dc982778ed2457d7b1 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ce78e1a1b0bea193650568ab12b3ff828391d624","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-16 12:45:49.000000000","message":"Patch Set 3: Code-Review+2\n\nTagging has gone through. Will leave to Thierry to approve once the missing-releases check has passed.","accounts_in_message":[],"_revision_number":3},{"id":"e9d72a180114a44f6e1a4b4a65bd59fc435388e6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-16 13:45:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d7a136d5d2a56bdafe874774c6e78e261ac6e5d7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-16 13:45:23.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"40f9ed55eb60e7d99d77766e23e53da5dd64fcf3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-16 13:45:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a77e8b0777f35a19d58ab667964a532c53ad43ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 13:45:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5dbf619bd4d8e08eb67b2e1a6770b3b52aaa5291","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-10-16 13:45:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d6568d31ffdabbef10903b50c211c5868952af7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:00:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ff9b7fda408548468f3db92ddd48eb074dc3f599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:00:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48052e4b2eb348aab8a03c0a7ec990f8 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f1f6abdce054113ab32f80a4efee9c4 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86097216939649a8aabdff05020886e4 : SUCCESS in 9m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec1811b36a16472d9c948e27d63a3dd5 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"807813eda8f873052177c8b65724b1fa3926957b","revisions":{"007fea60c0ce07f9b57dc92a05690874b58c9b4a":{"kind":"REWORK","_number":1,"created":"2019-10-10 22:06:36.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/687992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/687992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/687992/1"}}},"commit":{"parents":[{"commit":"b49733a34f89d6e0b64c07fccf1e05dc900af266","subject":"Train final releases for cycle-with-rc projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b49733a34f89d6e0b64c07fccf1e05dc900af266"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-10 22:06:04.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-10 22:06:04.000000000","tz":-300},"subject":"Mark Train as released","message":"Mark Train as released\n\nThe train has entered the station.\n\nChange-Id: I93747459c5e4555ea3e97a02bb49a2f3e400d6d1\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/007fea60c0ce07f9b57dc92a05690874b58c9b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/007fea60c0ce07f9b57dc92a05690874b58c9b4a"}]},"branch":"refs/heads/master"},"b1803ebf7c02562be625f137c089bc018fc0f4e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-15 07:15:27.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/92/687992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/687992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/687992/2"}}},"commit":{"parents":[{"commit":"fdb4b2cc9dbf99caea8df295ad66ecd7cfe424cf","subject":"Train final releases for cycle-with-rc projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fdb4b2cc9dbf99caea8df295ad66ecd7cfe424cf"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-10 22:06:04.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-10-15 06:55:20.000000000","tz":480},"subject":"Mark Train as released","message":"Mark Train as released\n\nThe train has entered the station.\n\nChange-Id: I93747459c5e4555ea3e97a02bb49a2f3e400d6d1\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1803ebf7c02562be625f137c089bc018fc0f4e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1803ebf7c02562be625f137c089bc018fc0f4e4"}]},"branch":"refs/heads/master"},"807813eda8f873052177c8b65724b1fa3926957b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-15 13:43:41.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/92/687992/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/92/687992/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/92/687992/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/92/687992/3"}}},"commit":{"parents":[{"commit":"86a537ba238364e07e43b04a5ccfbea323c29868","subject":"Train final releases for cycle-with-rc projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86a537ba238364e07e43b04a5ccfbea323c29868"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-10 22:06:04.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-15 13:42:02.000000000","tz":-300},"subject":"Mark Train as released","message":"Mark Train as released\n\nThe train has entered the station.\n\nChange-Id: I93747459c5e4555ea3e97a02bb49a2f3e400d6d1\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/807813eda8f873052177c8b65724b1fa3926957b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/807813eda8f873052177c8b65724b1fa3926957b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
