)]}'
{"id":"openstack%2Freleases~688407","triplet_id":"openstack%2Freleases~master~Iab182666ea65cdb7ea7e023853b18eaa9c57403b","project":"openstack/releases","branch":"master","topic":"train-final","hashtags":[],"change_id":"Iab182666ea65cdb7ea7e023853b18eaa9c57403b","subject":"Add Ussuri deliverable files","status":"MERGED","created":"2019-10-14 13:53:54.000000000","updated":"2019-10-22 14:41:02.000000000","submitted":"2019-10-22 14:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688407-1571755262589-9e4bfd77","meta_rev_id":"07631070c873e695e8ecb2389cf0702c2bdcd1c9","_number":688407,"virtual_id_number":688407,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 14:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-22 14:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-21 22:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-10-21 14:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-10-21 14:29:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-10-16 14:21:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2019-10-21 12:28:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-22 14:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 07:03:54.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-10-21 12:28:19.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-10-21 14:29:35.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-10-21 14:36:00.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-21 22:45:44.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-10-22 14:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4825c407e85451524a9e63c87a9e474c9ff194e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-14 13:53:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"039f5a35db3a4672a38866c64a50dc1d06a7ebcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 14:12:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce928dc3f1d4abfbd2ebd75a14219bd : FAILURE in 8m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f20000a14be4e22933777ad9166dfef : SUCCESS in 13m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f9885f73b54d4f4f8c93dc7f4afb3b78 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"75aa93952464453ae393569e6b4c237593b4ecd7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-14 14:49:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7deed615735703548aa356c0c2066ff95d895903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 15:08:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9c916eee4b42cdabb48af6f724a46e : SUCCESS in 9m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daa71032853a458fb17aa6263365744e : SUCCESS in 12m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8a316bbfa18047ce967b2031d62212fd : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ea0603e10c037207b004e9758cb1ba894a357d1f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-15 07:03:54.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"07c77ab1a586c2a6410c01c561b15782c54b17ea","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-15 07:15:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4c5c86b034b71931422577c04d99e3399caf2a1c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-15 07:24:15.000000000","message":"Topic novaclient-ussuri-16.0.0 removed","accounts_in_message":[],"_revision_number":3},{"id":"e048c5e5df9984c2f9e3d07a6ab394279cea1ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 07:57:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b799ad01603a46bda2f11a6150b8d740 : SUCCESS in 10m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6cfb2f4bc3724fb0b39467c98b4fa160 : SUCCESS in 14m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/445a5a8bfea34142b58dbdeb7399b931 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":3},{"id":"51649a2c4876e76c76b9908b52f8f8beb325ed41","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-15 13:43:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"773f45a0c7654992656f00b2b602bc31a6fbe3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:04:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44e71a9630a4a37b621084367b762a1 : SUCCESS in 9m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ca3847445014d539b89adefd67e8853 : SUCCESS in 13m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7ff135a1a90a46e692a8adcd7ad8f503 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":4},{"id":"afdfe216c7f369d21acd4f91fc6b6a65a5918802","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-16 14:21:47.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"577e0f3e56d5a68b81b81701d7e1af2a0c72514c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-16 14:24:50.000000000","message":"Topic changed from novaclient-ussuri-16.0.0 to train-final","accounts_in_message":[],"_revision_number":5},{"id":"144a28dc22c70c97c1c2ed8c259880be805ae602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 14:39:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89eef5ace874a63b71ca622a454d480 : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e26f83c1835d4f1682d9ea07432d9c4a : SUCCESS in 11m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ad24864ddfd6431ea0fcb421d9f154a1 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"4a58420c9ef20554a7f2db3e0588bfcd36a9a3aa","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-21 12:28:19.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"32b5e59e755f8a01828d3771612f61793665d8c5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-21 14:29:35.000000000","message":"Patch Set 5: Code-Review+1\n\nLooks good for nova, novaclient and os-vif.","accounts_in_message":[],"_revision_number":5},{"id":"f6c074bbc4faf78c8d26154fd36baaa3abce77fe","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-21 14:36:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5b034a8b3532acad0fff8d9e4abcbe2beb810628","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-21 22:45:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9f018ced3ef9fef6f41cc730e94ca106a610e272","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 14:21:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThis is largely procedural, and holding up some things, so pushing through.","accounts_in_message":[],"_revision_number":5},{"id":"f0e62729deaaa032efe2fe401a8079c5fbd2bdc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:21:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ce03747219b7aed766140a273484b26fd340cc51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:41:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcadedca09554ece891089554eff1e3d : SUCCESS in 9m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd15f86df26c4fa4a98c891d086f9ae2 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":5},{"id":"07631070c873e695e8ecb2389cf0702c2bdcd1c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 14:41:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f2559a011abd90eb89527c54b51ae5b42cba2241","revisions":{"6bab0691b7d24342c60415d7d864b6d506be12d9":{"kind":"REWORK","_number":1,"created":"2019-10-14 13:53:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/07/688407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/688407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/688407/1"}}},"commit":{"parents":[{"commit":"3a5cc27c68d16b26671651296680168157c2b91c","subject":"Merge \"pymod2pkg: Release 0.22.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a5cc27c68d16b26671651296680168157c2b91c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 13:52:13.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 13:52:13.000000000","tz":-300},"subject":"Add Ussuri deliverable files","message":"Add Ussuri deliverable files\n\nAdds deliverable stubs based on what was released in Train.\n\nChange-Id: Iab182666ea65cdb7ea7e023853b18eaa9c57403b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bab0691b7d24342c60415d7d864b6d506be12d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bab0691b7d24342c60415d7d864b6d506be12d9"}]},"branch":"refs/heads/master"},"4470561c9110a95599ead79b155ea502380bc34d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-14 14:49:39.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/07/688407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/688407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/688407/2"}}},"commit":{"parents":[{"commit":"007fea60c0ce07f9b57dc92a05690874b58c9b4a","subject":"Mark Train as released","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/007fea60c0ce07f9b57dc92a05690874b58c9b4a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 13:52:13.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 14:49:29.000000000","tz":-300},"subject":"Add Ussuri deliverable files","message":"Add Ussuri deliverable files\n\nAdds deliverable stubs based on what was released in Train.\n\nChange-Id: Iab182666ea65cdb7ea7e023853b18eaa9c57403b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4470561c9110a95599ead79b155ea502380bc34d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4470561c9110a95599ead79b155ea502380bc34d"}]},"branch":"refs/heads/master"},"b49240597e2962d0857d7f43a9cb18e625d51629":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-15 07:15:27.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/07/688407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/688407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/688407/3"}}},"commit":{"parents":[{"commit":"b1803ebf7c02562be625f137c089bc018fc0f4e4","subject":"Mark Train as released","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1803ebf7c02562be625f137c089bc018fc0f4e4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 13:52:13.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-10-15 06:55:20.000000000","tz":480},"subject":"Add Ussuri deliverable files","message":"Add Ussuri deliverable files\n\nAdds deliverable stubs based on what was released in Train.\n\nChange-Id: Iab182666ea65cdb7ea7e023853b18eaa9c57403b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b49240597e2962d0857d7f43a9cb18e625d51629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b49240597e2962d0857d7f43a9cb18e625d51629"}]},"branch":"refs/heads/master"},"1b17cec93755bb83b3ad43252df256713ff2a0ac":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-15 13:43:41.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/07/688407/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/688407/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/688407/4"}}},"commit":{"parents":[{"commit":"807813eda8f873052177c8b65724b1fa3926957b","subject":"Mark Train as released","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/807813eda8f873052177c8b65724b1fa3926957b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 13:52:13.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-15 13:42:43.000000000","tz":-300},"subject":"Add Ussuri deliverable files","message":"Add Ussuri deliverable files\n\nAdds deliverable stubs based on what was released in Train.\n\nChange-Id: Iab182666ea65cdb7ea7e023853b18eaa9c57403b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b17cec93755bb83b3ad43252df256713ff2a0ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b17cec93755bb83b3ad43252df256713ff2a0ac"}]},"branch":"refs/heads/master"},"f2559a011abd90eb89527c54b51ae5b42cba2241":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-10-16 14:21:47.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/07/688407/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/688407/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/688407/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/688407/5"}}},"commit":{"parents":[{"commit":"71656ee5cbc64ee06102b5cc66b93ce6047702a5","subject":"Merge \"Mark Train as released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71656ee5cbc64ee06102b5cc66b93ce6047702a5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-14 13:52:13.000000000","tz":-300},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2019-10-16 14:11:00.000000000","tz":480},"subject":"Add Ussuri deliverable files","message":"Add Ussuri deliverable files\n\nAdds deliverable stubs based on what was released in Train.\n\nChange-Id: Iab182666ea65cdb7ea7e023853b18eaa9c57403b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2559a011abd90eb89527c54b51ae5b42cba2241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2559a011abd90eb89527c54b51ae5b42cba2241"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
