)]}'
{"id":"openstack%2Freleases~689254","triplet_id":"openstack%2Freleases~master~I777a468c995c4f90852b9b14f6d188438357bc34","project":"openstack/releases","branch":"master","topic":"train-em","hashtags":[],"change_id":"I777a468c995c4f90852b9b14f6d188438357bc34","subject":"Fix Train extended maintenance transition date","status":"MERGED","created":"2019-10-17 19:49:02.000000000","updated":"2019-10-21 14:46:02.000000000","submitted":"2019-10-21 14:46:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689254-1571669162907-9e4bfd77","meta_rev_id":"986ee6525cf707003bf4e027709b822f6df0dbf2","_number":689254,"virtual_id_number":689254,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-21 14:46:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-10-21 14:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-10-21 13:24:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-19 00:22:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-10-21 14:34:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-19 00:22:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-21 13:24:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-10-21 14:34:36.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-21 14:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4e4d227c67e2f6fc8c60eae59f1e1a1825c71fd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-17 19:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7353af80bd1424021ec98fff710239ce537c2ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 20:04:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/230470c2515c441685844a186205d891 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b10b5da6fcdc4e1dafdce35b2c7ba3bc : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca928b547da4df681b83d1e4bc57b78 : SUCCESS in 10m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58d234ad1ff64e858522a38e096d93f9 : SUCCESS in 4m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5b57a15a47ce409cb6112007343831d1 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ede46bf5eb4b059ab4cf10203f4691 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c76714b1e15176c83cd88e1269a789d08b56cd41","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-19 00:22:06.000000000","message":"Patch Set 1: Code-Review+1\n\nI was just noticing this -- glad I looked for an existing patch before writing my own :-)","accounts_in_message":[],"_revision_number":1},{"id":"5364093243a6cd1d27b6ba0bb174a4cfc52aeb43","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-21 13:24:37.000000000","message":"Patch Set 1: Code-Review+1\n\nThe date is not yet an exact date (as it depends on the actual release date from master branch in 2021), but much better to have a more or less valid date than the current false date (2019-10-16) :)","accounts_in_message":[],"_revision_number":1},{"id":"9fe7c41d15d0dfa48514865bda6a2eb6bbfea568","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-21 14:34:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFast merging","accounts_in_message":[],"_revision_number":1},{"id":"3fa1399c3681497626ad733a8cdeed712e4145e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:34:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d002ae6a7662c75a4d7231a985b8dbb1dc9cf24c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:46:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ceaf3ba2783767214543337ecdf8401fdc3a820e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:46:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209691c5f7454736962243e98d557912 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b696cb1ecb8b42b4b97667ad7b049c7f : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67bce679dd924ecd82b4497d3daf018a : SUCCESS in 9m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26321d9a5d9f451ab8e9808e94edaa20 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1dbe53d46f48203c3c0274d2f49143b651deaff","revisions":{"b1dbe53d46f48203c3c0274d2f49143b651deaff":{"kind":"REWORK","_number":1,"created":"2019-10-17 19:49:02.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/54/689254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/689254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/689254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/689254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/689254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/689254/1"}}},"commit":{"parents":[{"commit":"997ab6144206003b685c135b20e5bc07b8693f6b","subject":"Release os-traits 1.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/997ab6144206003b685c135b20e5bc07b8693f6b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-17 19:48:04.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-17 19:48:04.000000000","tz":-300},"subject":"Fix Train extended maintenance transition date","message":"Fix Train extended maintenance transition date\n\nWe didn\u0027t actually want Train to go immediately into Extended\nMaintenance. This fixes the transition date for Train to be 18 months\nfrom its original release date.\n\nChange-Id: I777a468c995c4f90852b9b14f6d188438357bc34\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1dbe53d46f48203c3c0274d2f49143b651deaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b1dbe53d46f48203c3c0274d2f49143b651deaff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
