)]}'
{"id":"openstack%2Freleases~689786","triplet_id":"openstack%2Freleases~master~I68f050352eae3a5bc78cbeeaf37fc1190c0ef783","project":"openstack/releases","branch":"master","topic":"adopt-microversion-parse","hashtags":[],"change_id":"I68f050352eae3a5bc78cbeeaf37fc1190c0ef783","subject":"Add release history for microversion-parse","status":"MERGED","created":"2019-10-21 14:24:00.000000000","updated":"2019-10-30 10:16:41.000000000","submitted":"2019-10-30 10:16:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689786-1572430601067-edb63827","meta_rev_id":"d14e4be59aff1c3e4cd7289d0aac2a0732e8800b","_number":689786,"virtual_id_number":689786,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-10-30 09:45:15.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 10:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-21 16:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-22 18:46:14.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-22 18:48:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 18:46:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-10-22 18:48:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-30 10:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d005d8c139dd49fbb9be2f834e64a4b805e88d9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-21 14:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5d64c5773efd319cf8ae63cd8fe72f8ed4bc613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:44:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a62e09a04d42778dc12543d85bb635 : SUCCESS in 10m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3dc1e3f3d1f04416803a9654cb14e90e : FAILURE in 6m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c0e6f7f13eb14961ba6ed6f86f892d6d : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"74b387314e5af4409cb1ad0d95d0e312c6279965","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-21 15:01:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6642ac100535bcb49323b73823c3c72d27020d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 15:14:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f941145e44e43cc8967d840602dafa4 : SUCCESS in 9m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1319a84157ee4d4091c8979cdaf18430 : SUCCESS in 6m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4746d9b8cbe54a7c9a75afc57949dd6b : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d6da4e9d88edc32255a8d03ba53882e34e69d946","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-21 16:14:26.000000000","message":"Patch Set 2: Code-Review+2\n\nValidation is happy, so so am I. ;)","accounts_in_message":[],"_revision_number":2},{"id":"53b8ce4c2f54c9d8eba6d8cabae4f89ce35860fe","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-10-22 18:46:14.000000000","message":"Patch Set 2: Code-Review+2\n\nLike what Sean said.\n\nLooks good to me","accounts_in_message":[],"_revision_number":2},{"id":"e5cd5c363750b5499eb2e97ab5216f90b757b386","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-22 18:48:27.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c50fea60695bb9a17632cd3a2f1748adcf54aeb7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-30 09:45:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9d7d30f14e0cbc5c811db13901bb9c175e36b5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b172f38b174493924f368ac04576bd : SUCCESS in 9m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b300fa8202094e48853f23ba7a4109c7 : SUCCESS in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0d09071cbc254e38ae5bf4ce885d23e4 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"09148696f2c1326f467e9c4476f21930f76733c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:03:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2dd15c9763ef38f64fa8057f938438fa7fc9323","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:16:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d73dd98c3d4872a69499f0fb81fa07 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c128329c82542fbadb4e665dd138f4b : SUCCESS in 11m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18982be127b14421aec19c3cbe28b72c : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6206454644cbe1e1bfc93641a411f6363e140c5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:16:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bda749297f80e0ed8f696a904155a6182b519d50","revisions":{"b9d6d01011548f52703c8df52cd1fc64e44d73c5":{"kind":"REWORK","_number":1,"created":"2019-10-21 14:24:00.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/86/689786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/689786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/689786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/689786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/689786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/689786/1"}}},"commit":{"parents":[{"commit":"e4adf7b08e298c716bb46abdf6439a62996fc302","subject":"Merge \"TripleO RC tags + stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4adf7b08e298c716bb46abdf6439a62996fc302"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-10-21 14:22:47.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-10-21 14:22:47.000000000","tz":120},"subject":"Add release history for microversion-parse","message":"Add release history for microversion-parse\n\nmicroversion-parse is being adopted by the Oslo team.\nImport its release history.\n\nDepends-On: https://review.opendev.org/689754\nChange-Id: I68f050352eae3a5bc78cbeeaf37fc1190c0ef783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9d6d01011548f52703c8df52cd1fc64e44d73c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9d6d01011548f52703c8df52cd1fc64e44d73c5"}]},"branch":"refs/heads/master"},"bda749297f80e0ed8f696a904155a6182b519d50":{"kind":"REWORK","_number":2,"created":"2019-10-21 15:01:50.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/86/689786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/689786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/689786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/689786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/689786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/689786/2"}}},"commit":{"parents":[{"commit":"e4adf7b08e298c716bb46abdf6439a62996fc302","subject":"Merge \"TripleO RC tags + stable/train branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e4adf7b08e298c716bb46abdf6439a62996fc302"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-10-21 14:22:47.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-10-21 15:01:29.000000000","tz":120},"subject":"Add release history for microversion-parse","message":"Add release history for microversion-parse\n\nmicroversion-parse is being adopted by the Oslo team.\nImport its release history.\n\nDepends-On: https://review.opendev.org/689754\nChange-Id: I68f050352eae3a5bc78cbeeaf37fc1190c0ef783\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bda749297f80e0ed8f696a904155a6182b519d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bda749297f80e0ed8f696a904155a6182b519d50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
