)]}'
{"id":"openstack%2Freleases~691268","triplet_id":"openstack%2Freleases~master~Ia06ae76d3d030bab4c9c967d3789fd0a1786ae62","project":"openstack/releases","branch":"master","topic":"osc-sdk-bug-tracker","hashtags":[],"change_id":"Ia06ae76d3d030bab4c9c967d3789fd0a1786ae62","subject":"Update bug trackers of OSC and SDK","status":"MERGED","created":"2019-10-25 09:10:35.000000000","updated":"2019-10-25 12:10:18.000000000","submitted":"2019-10-25 12:10:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691268-1572005418802-b8c1e99e","meta_rev_id":"171bffc60cf3277d13578413cffd83ee9f68f3dd","_number":691268,"virtual_id_number":691268,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 12:10:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-25 12:01:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-25 12:01:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 12:01:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-25 12:10:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"858d5105d44e6325a3a3ed59fb7953ae808c618b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-25 09:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc1857d83f86e3a3f1582d3d538ea62a401f6b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 09:22:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4925308b57b8474abfece64c52df4ea0 : SUCCESS in 10m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22de52f7f8df49ce839abe571c0d3d21 : SUCCESS in 6m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7c1ebd8333714f98946bae3c6b3122d8 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"54cf5afc92733f584f7d1b5adae83d29978a2751","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 12:01:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"cabf4f348dde4d9e2601dc6645225609f8262eac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:01:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88d0e3641134ffcc5ef53f1a2b6c7ae8bba172b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:10:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a1998908b96df2e47937870af119dcb5e5fdaeb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:10:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2579b02a989b48cd8a1dcfc1a811c255 : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5e8718e08c5429081b28eed517f3404 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ca37129ebba051a71e99824816febce815a1c7a","revisions":{"6ca37129ebba051a71e99824816febce815a1c7a":{"kind":"REWORK","_number":1,"created":"2019-10-25 09:10:35.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/68/691268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/68/691268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/691268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/691268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/691268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/68/691268/1"}}},"commit":{"parents":[{"commit":"cd2230070c6aa47a8010d6be0a71912a0e28e8f2","subject":"Release openstacksdk 0.37.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd2230070c6aa47a8010d6be0a71912a0e28e8f2"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-25 08:08:12.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-25 08:10:23.000000000","tz":540},"subject":"Update bug trackers of OSC and SDK","message":"Update bug trackers of OSC and SDK\n\nOSC and SDK teams use storyboard as bug trackers.\nThis commit updates the bug tracker information.\n\nThe bug tracker of openstacksdk was wrong.\nIt pointed to openstack/shade (760). It is also fixed.\n\nChange-Id: Ia06ae76d3d030bab4c9c967d3789fd0a1786ae62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ca37129ebba051a71e99824816febce815a1c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ca37129ebba051a71e99824816febce815a1c7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
