)]}'
{"id":"openstack%2Freleases~691332","triplet_id":"openstack%2Freleases~master~I98533c3dd7ee3d1d2d785ab332938e1250fbb378","project":"openstack/releases","branch":"master","topic":"queens-em","hashtags":[],"change_id":"I98533c3dd7ee3d1d2d785ab332938e1250fbb378","subject":"Transition Kolla to Queens EM","status":"MERGED","created":"2019-10-25 14:04:41.000000000","updated":"2019-10-28 18:03:11.000000000","submitted":"2019-10-28 18:03:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691332-1572285791236-e84120e5","meta_rev_id":"963500486b7b1ac5d6ce9175e50ddcb25593045f","_number":691332,"virtual_id_number":691332,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 18:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-28 17:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-25 16:51:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-28 17:48:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 16:51:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-28 18:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5e07f72d1b29ae0ed4400836dff759c606474f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 14:04:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26cf6625133dd41c152df6980e38d7171a66b42f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:18:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fabedf3c2448c89c2efc184d027f9f : SUCCESS in 10m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7081b0d5da4c4fed9de9963c10b040ac : SUCCESS in 6m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/19621f93f077403d9febad52a9749368 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"85df3d3cd432129bcf3aeb0d48f1a2d133076f60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 14:49:57.000000000","message":"Patch Set 1: Code-Review-1\n\nHow flexible is the EM release deadline? We\u0027re a cycle-trailing project, and depend on the releases of other projects. I proposed a patch to kolla to bump the versions of packages for the final queens releases (https://review.opendev.org/#/c/691299/), which should be landing soon. Ideally we\u0027d cut a release including that change, then mark it as EM.\n\nSorry for not bringing this up sooner on the ML thread.","accounts_in_message":[],"_revision_number":1},{"id":"fc64d69ef0601eec351a076b131a23aea2d91a45","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 16:20:34.000000000","message":"Patch Set 1:\n\nNo worries Mark. We can hold off a bit so things can be wrapped up cleanly.","accounts_in_message":[],"_revision_number":1},{"id":"0195b883a0f4452ab7137c7b9ee9735cf84b4e6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 16:49:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1de62deee394a25a5189478b23970988d1aea174","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 16:51:29.000000000","message":"Patch Set 2:\n\nAdded https://review.opendev.org/#/c/691395 for final kolla releases, and rebased this patch onto it.","accounts_in_message":[],"_revision_number":2},{"id":"c81f80a46ca06fb4a9c0e7f8938b0549c938fab8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 16:51:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"67adc1f6c3a7e8bc396cc5abff9b0f7d167c1519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:05:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3b7944151840baa3c0a8c107e74bb0 : SUCCESS in 9m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/494549de255747fe8353207645b8cf92 : FAILURE in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f8cd5d07a4564e328139e28a3778937b : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e3045a7099607e2d7bb7f24d95d7c9ed29246bd9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 14:50:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fb9a811b53ac126cd1c81ce871c27c9f2a03493c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 15:00:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ab1aabf4e3488ebe9611551d4d0af6 : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e88d9799603f46098f53efca363d9fbd : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e27728a3a2874dc68aef8558ad1d54e1 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"16e37429d2e1a38a20b974de4eba9c8eb88c4465","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 17:48:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d29dd77ae146c2388e0fa6df4f994adc97acb68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:48:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf582533eb676aa66469c625908f5e73df10de03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:03:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88433c3bb4742f1ba1aca0c9a5282c4 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef93793e4ea4c119416c69cdd8f4173 : SUCCESS in 9m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11388dd8eb8f472a97d0ed99101bef7c : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b38f872f3b28452ddcc1758eeb00cca2000503cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:03:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c80b8d39344c636e6956aeece9a17f7821a60fc3","revisions":{"5926644544dc892f58bd8f38e58200607327bab2":{"kind":"REWORK","_number":1,"created":"2019-10-25 14:04:41.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/32/691332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/32/691332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/691332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/691332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/691332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/32/691332/1"}}},"commit":{"parents":[{"commit":"59e34a6c690864a694e186b0c5a6cdc83e5c6975","subject":"Merge \"Release ovsdbapp 0.12.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59e34a6c690864a694e186b0c5a6cdc83e5c6975"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 14:00:04.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 14:00:04.000000000","tz":-300},"subject":"Transition Kolla to Queens EM","message":"Transition Kolla to Queens EM\n\nChange-Id: I98533c3dd7ee3d1d2d785ab332938e1250fbb378\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5926644544dc892f58bd8f38e58200607327bab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5926644544dc892f58bd8f38e58200607327bab2"}]},"branch":"refs/heads/master"},"c80b8d39344c636e6956aeece9a17f7821a60fc3":{"kind":"REWORK","_number":2,"created":"2019-10-25 16:49:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/691332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/32/691332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/691332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/691332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/32/691332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/32/691332/2"}}},"commit":{"parents":[{"commit":"2a5d41846bc3771f6b7888e110227123e6580429","subject":"Release kolla 6.2.3 and kolla-ansible 6.2.4 for Queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a5d41846bc3771f6b7888e110227123e6580429"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 14:00:04.000000000","tz":-300},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-25 16:49:45.000000000","tz":60},"subject":"Transition Kolla to Queens EM","message":"Transition Kolla to Queens EM\n\nChange-Id: I98533c3dd7ee3d1d2d785ab332938e1250fbb378\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c80b8d39344c636e6956aeece9a17f7821a60fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c80b8d39344c636e6956aeece9a17f7821a60fc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
