)]}'
{"id":"openstack%2Freleases~691336","triplet_id":"openstack%2Freleases~master~I1543981cfba56abf59e98df1a8d64d5078e8b317","project":"openstack/releases","branch":"master","topic":"queens-em","hashtags":[],"change_id":"I1543981cfba56abf59e98df1a8d64d5078e8b317","subject":"Transition Mistral to Queens EM","status":"MERGED","created":"2019-10-25 14:04:49.000000000","updated":"2019-10-30 17:09:37.000000000","submitted":"2019-10-30 17:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691336-1572455377135-34f7ffe6","meta_rev_id":"107cd7057f304d1041ec4469c6805e57dc5c44f8","_number":691336,"virtual_id_number":691336,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 17:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-30 17:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-30 17:00:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 14:30:46.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-10-29 10:48:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-30 17:00:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-30 17:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b64b860d00b98e8e1e4b17d74b66fc25e4aa2b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 14:04:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"049173c5f0483d92ae131fa0bbdf99dfe80788e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:23:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbde38b0a474843be92be3a2a40ef2c : SUCCESS in 9m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b457d7fcd0134cd48ef2db6b3246940d : SUCCESS in 6m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/98409e02b3134440b4c347684b8d3bff : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a8957ac0a7ffc29a604b1da0ff04988e9e104160","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 17:28:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f4a485d8e684c593c56421d0c01c59faa42b206","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 17:29:30.000000000","message":"Patch Set 2:\n\nNeeded to break out some deliverables since our validation now enforces that all repos included in one deliverable are always released together.","accounts_in_message":[],"_revision_number":2},{"id":"35608aa4223a8a38f1ba07252bebf461a8135727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:39:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5746b41feb8d4f4f92bde313ba32288e : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bab4b3b8836d433d9c91f653c3ae6053 : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d60e64c5baee4f90a49736955ee10f12 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e6c9e7594efaecac5fa77654392d31208cca5cf3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-29 10:48:27.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"53eea170acfc188e07e2e44c6fce6dbe1dba0c2e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-30 14:40:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"130ef802ce94415c8719502c036c2aa3c7f96a64","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-30 14:41:36.000000000","message":"Patch Set 2: -Code-Review\n\nHmm, shouldn\u0027t mistral-dashboard and mistral-extra also have a queens-em tag?","accounts_in_message":[],"_revision_number":2},{"id":"39ab0b386aa7dacfc37bf1f6252e9da6ef4b1768","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 14:43:03.000000000","message":"Patch Set 2:\n\n\u003e Hmm, shouldn\u0027t mistral-dashboard and mistral-extra also have a\n \u003e queens-em tag?\n\nOh, good catch. They had to be broken out since we now enforce all repos in a single deliverable are released together and some of those were not. But after breaking them out, they didn\u0027t get the new em tag. I will update this.","accounts_in_message":[],"_revision_number":2},{"id":"7e1bd8e40d652ba1b749bdeb78ecb86de44a093f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 14:44:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1dd2c92dbf0a9326d7c5363bc691c210e4307ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 15:00:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0048ff16503e43c98b86c06166020e62 : SUCCESS in 11m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0997ccc52acc4c0393db6fed6266e9cf : SUCCESS in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/44caa3ef271447a38c0c6b397274f2f6 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3},{"id":"8dfaca33aca433c8f22c9ee4aa1551fd6ce21238","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-30 17:00:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5dc530a40fe7185931e2a4f663238b88c018da6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 17:00:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e5cb74539df02167f857b35327529ecf89394f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 17:09:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c345f90366be42cf871f2968f1da6a43 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ba370bf4be45228aa6d74c94517f45 : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84b062cc993f4323adecb7a3a28e6bae : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"e87061bb42a92e40f27d73678ce5bf3019e83305","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 17:09:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04282543cd6bd3a9ffd9ea1c7cffd78b8df7c23a","revisions":{"dea3637d92d78aa40d79b1bdcf2fe02fa0267584":{"kind":"REWORK","_number":1,"created":"2019-10-25 14:04:49.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/691336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/691336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/691336/1"}}},"commit":{"parents":[{"commit":"59e34a6c690864a694e186b0c5a6cdc83e5c6975","subject":"Merge \"Release ovsdbapp 0.12.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59e34a6c690864a694e186b0c5a6cdc83e5c6975"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:43:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:43:51.000000000","tz":-300},"subject":"Transition Mistral to Queens EM","message":"Transition Mistral to Queens EM\n\nChange-Id: I1543981cfba56abf59e98df1a8d64d5078e8b317\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dea3637d92d78aa40d79b1bdcf2fe02fa0267584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dea3637d92d78aa40d79b1bdcf2fe02fa0267584"}]},"branch":"refs/heads/master"},"a3a93f90cfa30ac47a01add789eb9234568911aa":{"kind":"REWORK","_number":2,"created":"2019-10-25 17:28:30.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/691336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/691336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/691336/2"}}},"commit":{"parents":[{"commit":"59e34a6c690864a694e186b0c5a6cdc83e5c6975","subject":"Merge \"Release ovsdbapp 0.12.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59e34a6c690864a694e186b0c5a6cdc83e5c6975"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:43:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 17:26:09.000000000","tz":-300},"subject":"Transition Mistral to Queens EM","message":"Transition Mistral to Queens EM\n\nChange-Id: I1543981cfba56abf59e98df1a8d64d5078e8b317\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3a93f90cfa30ac47a01add789eb9234568911aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3a93f90cfa30ac47a01add789eb9234568911aa"}]},"branch":"refs/heads/master"},"04282543cd6bd3a9ffd9ea1c7cffd78b8df7c23a":{"kind":"REWORK","_number":3,"created":"2019-10-30 14:44:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/691336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/691336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/691336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/691336/3"}}},"commit":{"parents":[{"commit":"59e34a6c690864a694e186b0c5a6cdc83e5c6975","subject":"Merge \"Release ovsdbapp 0.12.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59e34a6c690864a694e186b0c5a6cdc83e5c6975"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:43:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-30 14:44:35.000000000","tz":-300},"subject":"Transition Mistral to Queens EM","message":"Transition Mistral to Queens EM\n\nChange-Id: I1543981cfba56abf59e98df1a8d64d5078e8b317\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04282543cd6bd3a9ffd9ea1c7cffd78b8df7c23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04282543cd6bd3a9ffd9ea1c7cffd78b8df7c23a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
