)]}'
{"id":"openstack%2Freleases~691337","triplet_id":"openstack%2Freleases~master~I28acbc9ce13c74f4b154bbd2a5a334afea4d4a7e","project":"openstack/releases","branch":"master","topic":"queens-em","hashtags":[],"change_id":"I28acbc9ce13c74f4b154bbd2a5a334afea4d4a7e","subject":"Transition Monasca to Queens EM","status":"MERGED","created":"2019-10-25 14:04:51.000000000","updated":"2019-10-28 18:04:10.000000000","submitted":"2019-10-28 18:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691337-1572285850584-cc9bba01","meta_rev_id":"d55095b013a2751b2e24eb715bb1c14ca425e7d9","_number":691337,"virtual_id_number":691337,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 18:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-28 14:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-28 14:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-28 17:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 14:48:40.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-10-28 18:04:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b51d271d9096e53c05af379726015c25e34d7c02","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 14:04:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67f1d0c392891ccddc27a50dd73676a50fd58956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:26:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ec5bac6f9a46fb9cf0d6c1535c1ac9 : SUCCESS in 11m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a841a705c5664c81a29ea2c594c9c595 : SUCCESS in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c84e98c3687c4a39947ce2d1564cd18a : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b83feafd5ffad0063b04b3f805938f2a9996363c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-25 15:44:32.000000000","message":"Patch Set 1: Code-Review-1\n\nIf still possible I\u0027d like to include https://review.opendev.org/691377 and https://review.opendev.org/691378 as well. Sorry for the late request.","accounts_in_message":[],"_revision_number":1},{"id":"794d3ec509e8a95454faa17fff452f4035ca50d1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 11:34:06.000000000","message":"Patch Set 1:\n\n\u003e If still possible I\u0027d like to include https://review.opendev.org/691377\n \u003e and https://review.opendev.org/691378 as well. Sorry for the late\n \u003e request.\n\nPlease submit a release request ASAP.","accounts_in_message":[],"_revision_number":1},{"id":"1af96ce07a9c94fecb00919c2465d9bcb5be9cf1","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-28 12:18:58.000000000","message":"Patch Set 1:\n\nThe URLs point to the release requests.","accounts_in_message":[],"_revision_number":1},{"id":"8672ba35b23b894642aa8b73f9053a95c5e17efe","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-28 14:47:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40eed63aa68ba24294bd35052dc050a6559bbbe5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 14:47:39.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks Witold.","accounts_in_message":[],"_revision_number":2},{"id":"78d6dad5a0ed03a396324402c3f11af4a8a1ec13","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-28 14:48:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fcbd43969b3008618bfaceb642e1d88792deca78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:56:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41155f2a579f41379389bde87a8a6e33 : SUCCESS in 8m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9faf3ec4af854e7f983d0a9ba3b87644 : SUCCESS in 5m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fab38caf67814510a3c41b0b3319d36f : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cead1f3b5cbebfe152aa6236166b1e6adb2eb7d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 17:48:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d279851e5655af97a084575054dde7cf9554e076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:48:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ad17c197d8941087550ae9d5096793946ac2a26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:04:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1b34746fd166204dc6b7c360dce46d4e261be049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 18:04:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f768fcb0dd4f0c91a0508f1de4997d : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c3ab6c274c40c7b5edcc3eb722a73d : SUCCESS in 10m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/243deda556ad4904b4771519a2d1d8ad : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83a55ae0c258fbe6d624316059f3cad1524d092c","revisions":{"1a266c290a7aa5076af60b582b8f43d4091d1db1":{"kind":"REWORK","_number":1,"created":"2019-10-25 14:04:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/37/691337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/691337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/691337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/691337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/691337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/691337/1"}}},"commit":{"parents":[{"commit":"59e34a6c690864a694e186b0c5a6cdc83e5c6975","subject":"Merge \"Release ovsdbapp 0.12.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59e34a6c690864a694e186b0c5a6cdc83e5c6975"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:42:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:42:57.000000000","tz":-300},"subject":"Transition Monasca to Queens EM","message":"Transition Monasca to Queens EM\n\nChange-Id: I28acbc9ce13c74f4b154bbd2a5a334afea4d4a7e\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a266c290a7aa5076af60b582b8f43d4091d1db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a266c290a7aa5076af60b582b8f43d4091d1db1"}]},"branch":"refs/heads/master"},"83a55ae0c258fbe6d624316059f3cad1524d092c":{"kind":"REWORK","_number":2,"created":"2019-10-28 14:47:07.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/37/691337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/691337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/691337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/691337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/691337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/691337/2"}}},"commit":{"parents":[{"commit":"23606eea8d4e07a0a271e35c3eb17252861abe71","subject":"Merge \"Release monasca-ui 1.12.2 for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23606eea8d4e07a0a271e35c3eb17252861abe71"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:42:57.000000000","tz":-300},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-10-28 14:46:19.000000000","tz":60},"subject":"Transition Monasca to Queens EM","message":"Transition Monasca to Queens EM\n\nChange-Id: I28acbc9ce13c74f4b154bbd2a5a334afea4d4a7e\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83a55ae0c258fbe6d624316059f3cad1524d092c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83a55ae0c258fbe6d624316059f3cad1524d092c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
