)]}'
{"id":"openstack%2Freleases~691344","triplet_id":"openstack%2Freleases~master~I7d23ce344fc48e8f6fead367314327de61ea7925","project":"openstack/releases","branch":"master","topic":"queens-em","hashtags":[],"change_id":"I7d23ce344fc48e8f6fead367314327de61ea7925","subject":"Transition Puppet to Queens EM","status":"MERGED","created":"2019-10-25 14:05:04.000000000","updated":"2019-11-14 16:40:38.000000000","submitted":"2019-11-14 16:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691344-1573749638769-ee4c743d","meta_rev_id":"b004b05f9c01d7d9a4d600d1fc58c16375b4ddf5","_number":691344,"virtual_id_number":691344,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 16:40:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2019-11-14 03:36:40.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-11-12 11:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-11-07 14:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-11-14 16:29:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-10-25 14:51:42.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 14:51:42.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-11-07 14:34:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-14 03:36:40.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-11-14 16:29:57.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-14 16:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e8b12b23d724204bf2ddd9e011f4520411bc24b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 14:05:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dc04eef7b5cf188c0e74841fc1bda0bd328e875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0b0ae14d384cbbb487875e2790d0d0 : SUCCESS in 9m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a83f3b47f97d466b86c9152a3a08b070 : SUCCESS in 7m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b85b5ba6c53943b492137cb2e1062209 : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b497f34240e4801b0ba2ef03f295b80faa00bc7a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-25 14:56:24.000000000","message":"Patch Set 1: Code-Review-1\n\nplease hold off on this for now, not certain we want to EM this or keep it for a bit longer","accounts_in_message":[],"_revision_number":1},{"id":"603f625bf5524f179271b05237f25c08042a4cfb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-06 14:36:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23212ee1fa364b021647cde66ef8562edcf3f9a2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-07 14:34:14.000000000","message":"Patch Set 1: Code-Review+1\n\nper the tripleo review, as long as the publish items don\u0027t get run thent his should be fine.","accounts_in_message":[],"_revision_number":1},{"id":"f3d08da24fa50e04c61030a8c95578602ff10932","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-12 11:19:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e81c727ba43e19c95ab36ead736258cf42443019","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-14 03:34:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e9d53c48f098dc385d8f47bd81cb08731a9d5ce","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-14 03:36:40.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me.\n\nNot +W\u0027ing as I want to be certain about the tagging/artifacts/publishing","accounts_in_message":[],"_revision_number":1},{"id":"a4aec6507dd4e2a59ec2c829a7ad79df654c34b1","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-14 16:29:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67b5d8539907083bd7da28d8df8128cb94ff122c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a244e0835cd94db246a618a7989f5dba3ccac2ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:40:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59e3626499e5354d6b9db0325bb20faa3571fc0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:40:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f690b29962442728033967ab30e1b36 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1590ff9327ee4ac097dc0029760d7536 : SUCCESS in 9m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9ef1921073f4f77898ad45f27ebed4b : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c86b105841e8f9f5e660ce496a434be5de4e859","revisions":{"1c86b105841e8f9f5e660ce496a434be5de4e859":{"kind":"REWORK","_number":1,"created":"2019-10-25 14:05:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/44/691344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/691344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/691344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/691344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/691344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/691344/1"}}},"commit":{"parents":[{"commit":"59e34a6c690864a694e186b0c5a6cdc83e5c6975","subject":"Merge \"Release ovsdbapp 0.12.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/59e34a6c690864a694e186b0c5a6cdc83e5c6975"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:53:23.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 13:53:23.000000000","tz":-300},"subject":"Transition Puppet to Queens EM","message":"Transition Puppet to Queens EM\n\nChange-Id: I7d23ce344fc48e8f6fead367314327de61ea7925\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c86b105841e8f9f5e660ce496a434be5de4e859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c86b105841e8f9f5e660ce496a434be5de4e859"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
