)]}'
{"id":"openstack%2Freleases~691357","triplet_id":"openstack%2Freleases~master~I0ceb8d2b4394fdf90fc2bca230e0014aad5073f8","project":"openstack/releases","branch":"master","topic":"queens-em","hashtags":[],"change_id":"I0ceb8d2b4394fdf90fc2bca230e0014aad5073f8","subject":"Transition Watcher to Queens EM","status":"MERGED","created":"2019-10-25 14:05:30.000000000","updated":"2019-10-29 01:26:04.000000000","submitted":"2019-10-29 01:26:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691357-1572312364512-edb63827","meta_rev_id":"2e3e9860b411fad4ae7c3271f9885ecabb0d6b30","_number":691357,"virtual_id_number":691357,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 01:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-29 01:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-29 01:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-29 01:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 01:14:12.000000000","updated_by":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"reviewer":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"state":"REVIEWER"},{"updated":"2019-10-29 01:26:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"097dce6661a7db5aea661868dd50f19c29cad697","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 14:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f46d0bfdcaa5172c10f16c3c4cbf0634e4fff51f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:23:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3c2fe60a404bf0bc46fe437f0830ab : SUCCESS in 10m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce4ce1a9e7d943d88558c9e747177f34 : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/74b97445366a46d895ed41b54d69e240 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"70b8404ba6fbca448cc434ff33decd4ee28ef6af","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-10-28 01:40:31.000000000","message":"Patch Set 1: Code-Review-1\n\nI wish it includes the latest commit 5307f5a80ea81c69a1bc37aead007b441a4af450. thanks","accounts_in_message":[],"_revision_number":1},{"id":"49ec9906811295fb2d35cc913a74a1a0aee2bc80","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 10:41:25.000000000","message":"Patch Set 1:\n\nI have not seen a release request. We sent multiple posts the the mailing list to get any final ones done before the deadline. Was that missed?\n\nPlease request a release ASAP if you want to include any merged commits since the last release.","accounts_in_message":[],"_revision_number":1},{"id":"a89cc784e4236189835ef41601452a517e298412","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-10-28 11:53:51.000000000","message":"Patch Set 1:\n\n\u003e I have not seen a release request. We sent multiple posts the the\n \u003e mailing list to get any final ones done before the deadline. Was\n \u003e that missed?\n \u003e \n \u003e Please request a release ASAP if you want to include any merged\n \u003e commits since the last release.\nsorry...\nplease review https://review.opendev.org/#/c/691660/","accounts_in_message":[],"_revision_number":1},{"id":"a98a5dbfaf84a9cdbbdf4c129e209b4c33ed8a39","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 14:48:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66ecef71a0e76854babf698aee828529b73872f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 15:02:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e42d77ae474f199d5aec8790475167 : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f024e153c2f49818755d3b34900c9c0 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d88acfb1df2f4d2f96bc19545c132a34 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"86939b6af8709c65a53243853480d6c1e24bda12","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-10-29 01:14:12.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks!","accounts_in_message":[],"_revision_number":2},{"id":"da4d9a14df9a850c4282579303ad15ba0adba1c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-29 01:17:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"34a154f9b30ae73676488c5ced0d7ca3ce9c0c61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:17:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"955a7449f8adc321a7d627a0eb35e1913614e975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:26:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e13893516644ce5803af8ad717152a0 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04bcc769baa240899c5004cd399ebbfa : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1e7d567fc1648808a1d710ebc98cb18 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"89d8ac846f736a4b3fa0bd8dee6e7c28cef3a354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:26:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33a38929ebc8ab1adb6f9b3d823c075a18f310eb","revisions":{"ebb71c88163a6cc9684242101a11b8d2b465c3ad":{"kind":"REWORK","_number":1,"created":"2019-10-25 14:05:30.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/57/691357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/691357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/691357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/691357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/691357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/691357/1"}}},"commit":{"parents":[{"commit":"6ca37129ebba051a71e99824816febce815a1c7a","subject":"Update bug trackers of OSC and SDK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ca37129ebba051a71e99824816febce815a1c7a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 12:37:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 12:37:08.000000000","tz":-300},"subject":"Transition Watcher to Queens EM","message":"Transition Watcher to Queens EM\n\nChange-Id: I0ceb8d2b4394fdf90fc2bca230e0014aad5073f8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ebb71c88163a6cc9684242101a11b8d2b465c3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ebb71c88163a6cc9684242101a11b8d2b465c3ad"}]},"branch":"refs/heads/master"},"33a38929ebc8ab1adb6f9b3d823c075a18f310eb":{"kind":"REWORK","_number":2,"created":"2019-10-28 14:48:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/57/691357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/57/691357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/691357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/691357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/57/691357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/57/691357/2"}}},"commit":{"parents":[{"commit":"56f75e00a48129a92b805db23b3b3ee776b63a16","subject":"Merge \"Watcher release queens 1.8.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56f75e00a48129a92b805db23b3b3ee776b63a16"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-25 12:37:08.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-10-28 14:48:40.000000000","tz":-300},"subject":"Transition Watcher to Queens EM","message":"Transition Watcher to Queens EM\n\nChange-Id: I0ceb8d2b4394fdf90fc2bca230e0014aad5073f8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33a38929ebc8ab1adb6f9b3d823c075a18f310eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33a38929ebc8ab1adb6f9b3d823c075a18f310eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
