)]}'
{"id":"openstack%2Freleases~691401","triplet_id":"openstack%2Freleases~master~I872ccd63bd4e1bfd197b9fc588113467f3e534b0","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I872ccd63bd4e1bfd197b9fc588113467f3e534b0","subject":"Final keystone releases for Queens","status":"MERGED","created":"2019-10-25 17:35:24.000000000","updated":"2019-10-30 14:10:58.000000000","submitted":"2019-10-30 14:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691401-1572444658043-34f7ffe6","meta_rev_id":"141351051831aff04b55659c006928cf2b6dfce6","_number":691401,"virtual_id_number":691401,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 14:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-28 14:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-28 16:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-10-30 13:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-10-30 13:57:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 14:29:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-28 16:54:17.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-10-30 13:57:56.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-10-30 14:10:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13b1086e0611fa129dc40d1c06ab789cfab03924","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-10-25 17:35:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81c059e3ed161403d85fac4559f27fd755a9e86d","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-10-25 17:36:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d37c9a120edef508846f0d320e095cda6d52ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:50:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a0c0357956426285552852353d71fd : SUCCESS in 8m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d975d38280dd4c5ab338576504d550f4 : SUCCESS in 7m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d4fc8b603874414189de1cc8b1742042 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7a143eef4b050f473e6d620ea7afc77301db5be3","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-10-25 18:05:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25b44d2fcf322d581fec50e91ff497dbe2086860","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-25 18:09:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"71a52be0e6a5a8dd0d32f610c3f7db5ed81f09bf","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-10-25 18:15:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9bd5b8c13af68df83fe8cb286f5c7aa7b1f328c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:32:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52128de5535c41258025434cd8f21018 : SUCCESS in 11m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8e90e48824c4704a6e85114128ca03e : SUCCESS in 8m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/03844e33598b420297aab58e350fa391 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":4},{"id":"c229c0b145a2f4f83af96c1fb4a560bce28b67a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 18:35:28.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"919c160e6f1f891d50b1dc9ac908ce4d6e20d72e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-25 18:52:38.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"394312ba1bd5e7b12f1702b221428ae5a00c7997","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-25 19:04:23.000000000","message":"Patch Set 4: Code-Review+2\n\nAh, I did miss those. Thanks Colleen.\n\nStill think the version number on keystonemiddleware should maybe be bumped, but not a huge deal. At least it appears the new config option default makes the behavior the same.","accounts_in_message":[],"_revision_number":4},{"id":"6ec433f5d08b871989034b975067731fe6463c74","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-25 19:10:54.000000000","message":"Patch Set 4:\n\n\u003e Still think the version number on keystonemiddleware should maybe\n \u003e be bumped, but not a huge deal. At least it appears the new config\n \u003e option default makes the behavior the same.\n\nYeah that change was backported as a bugfix but it is technically a feature, luckily the rocky series doesn\u0027t already have that tag so it would be fine to make that a minor bump, I don\u0027t have strong feelings either way.","accounts_in_message":[],"_revision_number":4},{"id":"2c1c24d3aa4b0bb25c96e4598b718012673d2d6a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-28 14:27:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"19e502546f11601b6b092ab5a6ec551d1e8e3287","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-28 14:29:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"10a7812e551e01d0b2ff1093b5975a63b208de7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:42:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3ed10e12814afea17857f77479a9f1 : SUCCESS in 9m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/124692cc636949b9bd8d71f94197aa4a : SUCCESS in 7m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/57cff5ec3736462fbaa57b75fed96e79 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":5},{"id":"929b162615c392890fe5a6c1f3d68b6b2c5b892d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-28 16:54:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"59541213222ee4205dd9d6f5c9fec17bd2f89376","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-30 13:57:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e87d335942992c0cb45df9a22ad9e533d291f8b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:58:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9906eacbcd000db5c917d0036ebea2f9d3b49ca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:10:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09dee32c8be4851b51bc97a9edb3f42 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44933616e86947e38c906b6f413b1c48 : SUCCESS in 9m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c6d2c5098794102ade7d277392a8fb1 : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":5},{"id":"141351051831aff04b55659c006928cf2b6dfce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:10:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7a7c617ceac87c7d239f9f068f1869ec71eb2c1a","revisions":{"47bcccb7adebdf78f0d25599adc7f15fbe6c9197":{"kind":"REWORK","_number":1,"created":"2019-10-25 17:35:24.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/01/691401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/691401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/691401/1"}}},"commit":{"parents":[{"commit":"eb9fb12a6afcbb07b2fbe920b484bdb139b02f02","subject":"Merge \"Transition Horizon to Queens EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9fb12a6afcbb07b2fbe920b484bdb139b02f02"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:34:40.000000000","tz":-240},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:34:40.000000000","tz":-240},"subject":"Release keystone 10.0.3","message":"Release keystone 10.0.3\n\nChange-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47bcccb7adebdf78f0d25599adc7f15fbe6c9197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47bcccb7adebdf78f0d25599adc7f15fbe6c9197"}]},"branch":"refs/heads/master"},"e9851dda26dfc2a4fe26167577da47aeaf955ef6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-25 17:36:10.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/01/691401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/691401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/691401/2"}}},"commit":{"parents":[{"commit":"eb9fb12a6afcbb07b2fbe920b484bdb139b02f02","subject":"Merge \"Transition Horizon to Queens EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9fb12a6afcbb07b2fbe920b484bdb139b02f02"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:34:40.000000000","tz":-240},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:35:52.000000000","tz":-240},"subject":"Release keystone 13.0.3","message":"Release keystone 13.0.3\n\nChange-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9851dda26dfc2a4fe26167577da47aeaf955ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9851dda26dfc2a4fe26167577da47aeaf955ef6"}]},"branch":"refs/heads/master"},"b7269197ef866c3710405d9a1ab6199fcfafd7a4":{"kind":"REWORK","_number":3,"created":"2019-10-25 18:05:20.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/01/691401/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/691401/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/691401/3"}}},"commit":{"parents":[{"commit":"eb9fb12a6afcbb07b2fbe920b484bdb139b02f02","subject":"Merge \"Transition Horizon to Queens EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9fb12a6afcbb07b2fbe920b484bdb139b02f02"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:34:40.000000000","tz":-240},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 18:04:40.000000000","tz":-240},"subject":"Final keystone releases for Queens","message":"Final keystone releases for Queens\n\nChange-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7269197ef866c3710405d9a1ab6199fcfafd7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7269197ef866c3710405d9a1ab6199fcfafd7a4"}]},"branch":"refs/heads/master"},"3d16463836613cb713998dc8749db7e3d10c2914":{"kind":"REWORK","_number":4,"created":"2019-10-25 18:15:03.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/01/691401/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/691401/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/691401/4"}}},"commit":{"parents":[{"commit":"eb9fb12a6afcbb07b2fbe920b484bdb139b02f02","subject":"Merge \"Transition Horizon to Queens EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9fb12a6afcbb07b2fbe920b484bdb139b02f02"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:34:40.000000000","tz":-240},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 18:14:18.000000000","tz":-240},"subject":"Final keystone releases for Queens","message":"Final keystone releases for Queens\n\nChange-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d16463836613cb713998dc8749db7e3d10c2914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3d16463836613cb713998dc8749db7e3d10c2914"}]},"branch":"refs/heads/master"},"7a7c617ceac87c7d239f9f068f1869ec71eb2c1a":{"kind":"REWORK","_number":5,"created":"2019-10-28 14:27:50.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/01/691401/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/691401/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/691401/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/691401/5"}}},"commit":{"parents":[{"commit":"eb9fb12a6afcbb07b2fbe920b484bdb139b02f02","subject":"Merge \"Transition Horizon to Queens EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9fb12a6afcbb07b2fbe920b484bdb139b02f02"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-10-25 17:34:40.000000000","tz":-240},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-10-28 14:27:43.000000000","tz":0},"subject":"Final keystone releases for Queens","message":"Final keystone releases for Queens\n\nChange-Id: I872ccd63bd4e1bfd197b9fc588113467f3e534b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a7c617ceac87c7d239f9f068f1869ec71eb2c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a7c617ceac87c7d239f9f068f1869ec71eb2c1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
