)]}'
{"id":"openstack%2Freleases~691772","triplet_id":"openstack%2Freleases~master~Ib29e8815e08b78607da3807dae1693ca8bb2efc4","project":"openstack/releases","branch":"master","topic":"manila","hashtags":[],"change_id":"Ib29e8815e08b78607da3807dae1693ca8bb2efc4","subject":"[manila] Release 9.1.0 off of stable/train","status":"MERGED","created":"2019-10-29 00:22:03.000000000","updated":"2019-11-12 11:19:54.000000000","submitted":"2019-11-12 11:19:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691772-1573557594139-ebba688c","meta_rev_id":"de71a0e4ee922ea71d4c711e7afa8b0d7b995aa6","_number":691772,"virtual_id_number":691772,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-10-29 17:12:19.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-10-29 09:54:26.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-12 11:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-29 17:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2019-11-12 11:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-10-31 15:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2019-11-12 11:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 09:54:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-29 17:13:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-31 15:07:45.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-11-12 11:09:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-12 11:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec6cafbdaf07e9da990f7977af7fb2e8cef46232","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-29 00:22:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1476271a2ced5b9fd3880b3aa8898cc520964d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 00:29:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec2d16913644ba481e72d810d14c517 : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d15234df0b24fcd9336519a278cb585 : SUCCESS in 5m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9fc2dc50c707463b854c30b34df76ff4 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"572376ddf1aa0ee06d5e816152e2ecdcf76db202","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-29 00:35:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6d0752b7cdd36081bc100b42abcc8ccaa6a62723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 00:42:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663fcf26473d4d53b2938ec46bfa0380 : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd362f6336204f95b0819bb0a9aa9ba3 : SUCCESS in 4m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4edde9e2e3f545bfb8bdfe8402dd52e2 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"10a913e81ad97b6989bf9b6514654e69aff68c38","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-29 09:54:26.000000000","message":"Patch Set 2:\n\nI only seen fixes in the embdded commits, so I guess it\u0027s more a 9.0.1 rather than a 9.1.0\n\nThoughts?","accounts_in_message":[],"_revision_number":2},{"id":"a237bcc9b83c589561afe4793d35caef427c54fb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-29 15:59:30.000000000","message":"Patch Set 2:\n\nCommit mentions the API changes, but I didn\u0027t see where. From the changes included, this does look more like it should be a 9.0.1 (though \"[NetApp] Allow extension/shrinking of NetApp replicated share\" is a pretty gray area for being an acceptable backport).","accounts_in_message":[],"_revision_number":2},{"id":"354bc114d38330c21c819de3de9fde8f97697a5f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-29 17:12:19.000000000","message":"Patch Set 2:\n\nHerve, Sean: \n\nThank you for taking a look! :)\n\nThe bugfix for https://bugs.launchpad.net/manila/+bug/1846836 changes a HTTP200 to HTTP400, which is usually taboo for us, but, we decided to land the change because it was terrible user experience - the same bug fix also prevents incorrect partial updates to share networks which were previously allowed.\n\nThe bugfix for https://bugs.launchpad.net/manila/+bug/1700871 is internal to the NetApp driver, and was to prevent a foobar situation with shares on that driver after a failover/failback. Do you see anything that doesn\u0027t conform to stable back port guidelines?","accounts_in_message":[],"_revision_number":2},{"id":"c60483d1674e3ddf5614fdf0fe476bcd3af4f8f1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-29 17:13:19.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e Herve, Sean:\n \u003e \n \u003e Thank you for taking a look! :)\n \u003e \n \u003e The bugfix for https://bugs.launchpad.net/manila/+bug/1846836\n \u003e changes a HTTP200 to HTTP400, which is usually taboo for us, but,\n \u003e we decided to land the change because it was terrible user\n \u003e experience - the same bug fix also prevents incorrect partial\n \u003e updates to share networks which were previously allowed.\n \u003e \n \u003e The bugfix for https://bugs.launchpad.net/manila/+bug/1700871 is\n \u003e internal to the NetApp driver, and was to prevent a foobar\n \u003e situation with shares on that driver after a failover/failback. Do\n \u003e you see anything that doesn\u0027t conform to stable back port\n \u003e guidelines?\n\nMakes sense. Thanks for the explanation Goutham.","accounts_in_message":[],"_revision_number":2},{"id":"c78e47142c856a21a7760413df0f4b5338f0299e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-10-30 10:20:13.000000000","message":"Patch Set 2:\n\nLGTM, pausing to give stable team a bit more time to review","accounts_in_message":[],"_revision_number":2},{"id":"d7e6976aa593f6a0c5155592aea4ab9444130e85","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-31 15:07:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"63d8febc8be21833aa4e06e4dbe42a9a3bf8af91","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-12 11:09:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"679ad7b5f05d6ea34f5a7e7e9b0ab90c24280784","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 11:09:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11248c164cd7d739cd56f95019a1d1394c4df09d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 11:19:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3792783e849645a19333707f4bf6562c : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d64d19d2d14ee69cc503ffa50726e9 : SUCCESS in 8m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/830df0f009e24e019fb8f263f579f753 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"687209bb3f2d6f4dc7d8d66aebe6f56d45887e9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 11:19:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6250c4764f37ce578bc7db41306708a1cfba8202","revisions":{"a7335fed9aa4dd93e557d25048f665fcbf33a058":{"kind":"REWORK","_number":1,"created":"2019-10-29 00:22:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/691772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/691772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/691772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/691772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/691772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/691772/1"}}},"commit":{"parents":[{"commit":"6bf8e81822be65c275ae4c5f67df2ee27a0075c7","subject":"Merge \"Categorize mistral-dashboard into horizon-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bf8e81822be65c275ae4c5f67df2ee27a0075c7"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-29 00:20:11.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-29 00:20:11.000000000","tz":-420},"subject":"[manila] Release 9.0.1 off of stable/train","message":"[manila] Release 9.0.1 off of stable/train\n\nThis minor release includes bug-fixes for:\n\n- bug #1846836\n- bug #1700871\n- bug #1634734\n\nChange-Id: Ib29e8815e08b78607da3807dae1693ca8bb2efc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7335fed9aa4dd93e557d25048f665fcbf33a058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7335fed9aa4dd93e557d25048f665fcbf33a058"}]},"branch":"refs/heads/master"},"6250c4764f37ce578bc7db41306708a1cfba8202":{"kind":"REWORK","_number":2,"created":"2019-10-29 00:35:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/691772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/691772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/691772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/691772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/691772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/691772/2"}}},"commit":{"parents":[{"commit":"6bf8e81822be65c275ae4c5f67df2ee27a0075c7","subject":"Merge \"Categorize mistral-dashboard into horizon-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bf8e81822be65c275ae4c5f67df2ee27a0075c7"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-29 00:20:11.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-10-29 00:35:03.000000000","tz":0},"subject":"[manila] Release 9.1.0 off of stable/train","message":"[manila] Release 9.1.0 off of stable/train\n\nThis minor release includes bug-fixes for:\n\n- bug #1846836\n- bug #1700871\n- bug #1634734\n\nThese bugfixes affect the public API for the Shared\nFile Systems service, but introduce no backwards\nincompatible changes that affect the client negatively.\n\nChange-Id: Ib29e8815e08b78607da3807dae1693ca8bb2efc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6250c4764f37ce578bc7db41306708a1cfba8202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6250c4764f37ce578bc7db41306708a1cfba8202"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
