)]}'
{"id":"openstack%2Freleases~692410","triplet_id":"openstack%2Freleases~master~I081366d33a49d6f73d63ca57af667b5e5222b473","project":"openstack/releases","branch":"master","topic":"keystone-13.0.4","hashtags":[],"change_id":"I081366d33a49d6f73d63ca57af667b5e5222b473","subject":"Release keystone 13.0.4 for queens","status":"MERGED","created":"2019-10-31 16:55:59.000000000","updated":"2019-10-31 20:39:45.000000000","submitted":"2019-10-31 20:39:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692410-1572554385545-8fb7e297","meta_rev_id":"53ee2c6330b529a97c662e8ed034b29511e4769d","_number":692410,"virtual_id_number":692410,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-31 20:39:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-31 20:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-31 20:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 20:27:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-31 20:39:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c907de8af0cc42683a21c91640207f62578f218","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-31 16:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"090838f28c6e2d0d881ed7553bfeb6d14f22183a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 17:05:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc538990ac346128cef74636e6eda15 : SUCCESS in 8m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e8949fa33754c38a96f7241749a73db : SUCCESS in 6m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1fa780c5e0194702a0f35e5ef5f21567 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"10c09510afd6f14ae0e7246594dc663894544f77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-31 20:27:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSince this is been well communicated and there\u0027s just one commit that addresses a big functionality usage issue, pushing this through.","accounts_in_message":[],"_revision_number":1},{"id":"6b63295d9283a03a5e515642a27ba80471d3701e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 20:27:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28acb03287db409fb8c51561a445e10a40664f50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 20:39:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a36fdb3f994f8b6e5d68d19fe2d5e5ac8fb7b489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 20:39:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906874080ad0445db3ff421ba4063586 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ff3258a51144868582abb1b8e96c54 : SUCCESS in 10m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a26db86042244cfac656d20b4d05e4f : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56f8e68fbccd2029c358b641b4e997f95a50e77e","revisions":{"56f8e68fbccd2029c358b641b4e997f95a50e77e":{"kind":"REWORK","_number":1,"created":"2019-10-31 16:55:59.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/10/692410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/10/692410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/692410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/692410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/692410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/10/692410/1"}}},"commit":{"parents":[{"commit":"9751f59a7c3aae9e7d77f8a12bfeb6d218c0187b","subject":"Merge \"Add Cinder-specific dates for Ussuri schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9751f59a7c3aae9e7d77f8a12bfeb6d218c0187b"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-31 16:55:45.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-31 16:55:45.000000000","tz":-420},"subject":"Release keystone 13.0.4 for queens","message":"Release keystone 13.0.4 for queens\n\nChange-Id: I081366d33a49d6f73d63ca57af667b5e5222b473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56f8e68fbccd2029c358b641b4e997f95a50e77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56f8e68fbccd2029c358b641b4e997f95a50e77e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
