)]}'
{"id":"openstack%2Freleases~693125","triplet_id":"openstack%2Freleases~master~Ib5d582dbe092a54ccf307d667cb405d13e2eb453","project":"openstack/releases","branch":"master","topic":"fix-kibana-link","hashtags":[],"change_id":"Ib5d582dbe092a54ccf307d667cb405d13e2eb453","subject":"Remove artifact links for monasca-kibana-plugin","status":"MERGED","created":"2019-11-06 06:33:53.000000000","updated":"2019-11-11 15:32:34.000000000","submitted":"2019-11-11 15:32:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693125-1573486354769-2b47506d","meta_rev_id":"b521787d6f43318ffc96f4950fc5fadba64e7a26","_number":693125,"virtual_id_number":693125,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-11 15:32:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-07 01:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-11-11 15:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 15:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-11 15:32:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ef8bb815a0218188864065760f3860d7af90e36","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-06 06:33:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10c0880be7c663afd5d2d85d9112e77e22f62c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 06:44:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d7bc625a4a541d28b706abb97f9b080 : SUCCESS in 8m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9931ccc1b3ce4e278a59278ad7bcebef : SUCCESS in 4m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8e1be9c4119a4c54a5858789a343bf99 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7c805d64232c59b37bebbbe342373896c3484df4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-07 01:40:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8daa4af7430e34315c5a8b4e743896390e85babd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-11 15:20:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90ca73da7f262e9e67b3f35b25eaee00063380fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:20:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c36c9839779362c1417f5f2dcb7c8b9b5df7005c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:32:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b2720f60ec4c009dce2317a40aa378 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab96511ef6ce477383baef74fb13b359 : SUCCESS in 9m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91b289f26e1948498bba36c3db7898e9 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"706e0fb89914aeb27551f8fa22d61dda57e40755","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:32:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3be70b0f9259f361fc155cd36eae173e5de59c8f","revisions":{"3be70b0f9259f361fc155cd36eae173e5de59c8f":{"kind":"REWORK","_number":1,"created":"2019-11-06 06:33:53.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/25/693125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/25/693125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/693125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/693125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/25/693125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/25/693125/1"}}},"commit":{"parents":[{"commit":"9106dd60c31c249dc604583046df6d696de4ee53","subject":"Merge \"remove dhellmann as release liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9106dd60c31c249dc604583046df6d696de4ee53"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-06 06:24:41.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-06 06:32:32.000000000","tz":60},"subject":"Remove artifact links for monasca-kibana-plugin","message":"Remove artifact links for monasca-kibana-plugin\n\nSet artifact-link-mode: none for monasca-kibana-plugin to avoid\nbroken pgp signature links. Retroactively fix Stein and Train.\n\nChange-Id: Ib5d582dbe092a54ccf307d667cb405d13e2eb453\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3be70b0f9259f361fc155cd36eae173e5de59c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3be70b0f9259f361fc155cd36eae173e5de59c8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
