)]}'
{"id":"openstack%2Freleases~694447","triplet_id":"openstack%2Freleases~master~I0d15b5137d857bffb2bed39e995e50fed24f5598","project":"openstack/releases","branch":"master","topic":"feature/new_tools","hashtags":[],"change_id":"I0d15b5137d857bffb2bed39e995e50fed24f5598","subject":"Add simple liaisons wrapper module","status":"MERGED","created":"2019-11-15 04:19:10.000000000","updated":"2020-01-09 18:41:53.000000000","submitted":"2020-01-09 18:41:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694447-1578595313426-0548d2f5","meta_rev_id":"20893082c072dc094d7fb76de3a48fdc7fb6c3a7","_number":694447,"virtual_id_number":694447,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 18:41:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-11-18 09:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-20 14:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-20 14:10:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 09:14:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-18 09:51:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-20 14:10:39.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-09 18:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1930765ea460846f1c0a83da53e251bf4984bf31","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-15 04:19:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"852ff880b4e88b9e72857789153978adf2b057e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 04:31:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c95b4f041143649d60f9418d3291d6 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/161b031fd30d4038a74a2fa530940840 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ab9aef690764ea3b77815414169a5ef : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7371baceea4dd6be7531102245842a : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdba915aabcb4d4b99c08a0a094dd011 : SUCCESS in 4m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/891f0ae86a174eff895ee9f2addd2e84 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"56731ed9e3b1bb0508bbfc5d04b2de0f7041e86e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-15 09:14:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b48ae94ca7c6e083a1e02e992eea48665d9a58fe","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-18 01:25:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcb8f081be1b2427820c7436d92991b980da547d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-18 01:34:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cff2a4572b4c14b888a188cabb002697daeed2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 01:48:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80df30c1bb3b44db8be77c048a5eb3c7 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7434f45d8d844208a46bfe38f5593030 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0ea76b385d54aafb05677a68182998b : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94ba0a247a74b95a7294a3790b2cc74 : SUCCESS in 8m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f456d507cff04faca4861111d57f64d9 : SUCCESS in 4m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/722e3d22de64472d8337ea7c727bbbb7 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b933a8cacc7963644897e6ae3eb51fc646c0607d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-18 09:51:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3b79bf5d18b0fa1b0c5eafba425c7bfd923d513a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-20 14:10:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0708aee7e5075f1d7f2a8b65a2c7001168373790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:57:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c39d7801ec05a0256cb9566839f3929a52418a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 18:41:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6fcf9a18a17302f4fe9061c76529199ccf44fa3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 18:41:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7927c1b0fb4e48c9a5d5c5f67db79af3 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/768c0a3825ff4906979237c0881d1a67 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60fe85ea07804a2eafc4934f008bc142 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f329c0875e74fd788a8d8ba9ef9dcf3 : SUCCESS in 9m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38d9f0e05b2f4e1fa58ba93fcd9db86d : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"390883c17d81f3b670ff0fe97cdb3ce0e613580e","revisions":{"6626b32fb338bc5f3c6d51af2961fd666904e33e":{"kind":"REWORK","_number":1,"created":"2019-11-15 04:19:10.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/47/694447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/694447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/694447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/694447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/694447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/694447/1"}}},"commit":{"parents":[{"commit":"d3ffb8d5273f18daf1a9055195ad2ab074f6c568","subject":"Add wrapper script that can be used to update deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3ffb8d5273f18daf1a9055195ad2ab074f6c568"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-14 23:21:32.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-15 04:03:41.000000000","tz":660},"subject":"Add simple liaisons wrapper module","message":"Add simple liaisons wrapper module\n\nI\u0027m going to want to use this in a later change so lets make the\nabstraction now\n\nChange-Id: I0d15b5137d857bffb2bed39e995e50fed24f5598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6626b32fb338bc5f3c6d51af2961fd666904e33e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6626b32fb338bc5f3c6d51af2961fd666904e33e"}]},"branch":"refs/heads/master"},"390883c17d81f3b670ff0fe97cdb3ce0e613580e":{"kind":"REWORK","_number":2,"created":"2019-11-18 01:34:38.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/47/694447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/694447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/694447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/694447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/694447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/694447/2"}}},"commit":{"parents":[{"commit":"7045ff71b0db13b00adf2159a6f89b33a17768ba","subject":"Add wrapper script that can be used to update deliverables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7045ff71b0db13b00adf2159a6f89b33a17768ba"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-14 23:21:32.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-18 01:30:59.000000000","tz":660},"subject":"Add simple liaisons wrapper module","message":"Add simple liaisons wrapper module\n\nI\u0027m going to want to use this in a later change so lets make the\nabstraction now\n\nChange-Id: I0d15b5137d857bffb2bed39e995e50fed24f5598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/390883c17d81f3b670ff0fe97cdb3ce0e613580e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/390883c17d81f3b670ff0fe97cdb3ce0e613580e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
