)]}'
{"id":"openstack%2Freleases~694786","triplet_id":"openstack%2Freleases~master~I2c0ca25ff960d5d969746df6a851253d4fa69faa","project":"openstack/releases","branch":"master","topic":"bug/1850843","hashtags":[],"change_id":"I2c0ca25ff960d5d969746df6a851253d4fa69faa","subject":"Feature release for Mistral Stein","status":"MERGED","created":"2019-11-18 14:37:52.000000000","updated":"2019-11-22 13:55:44.000000000","submitted":"2019-11-22 13:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694786-1574430944463-8d7acffc","meta_rev_id":"9d19062dfcff71d0b2d629b94e2d3c3473739215","_number":694786,"virtual_id_number":694786,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 13:55:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-21 16:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-11-22 13:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-11-21 10:21:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-11-22 13:41:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 14:08:19.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-11-21 10:21:23.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-21 16:33:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-22 13:41:26.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-22 13:55:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6f429006838dc70e0e20e90002677a24596b826","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 14:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dba0b9642ef5d5d406be11682b5c7783a7f0b77","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-18 14:38:31.000000000","message":"Topic set to bug/1850843","accounts_in_message":[],"_revision_number":1},{"id":"32c54fefbe53d71c17cd8f8464cd1cfbecdea248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:53:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3898d3c47ca24ba697a7ff47e54ae5a9 : SUCCESS in 10m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/165bc7bddae946ec9621bacbf80268a7 : SUCCESS in 5m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/32bd2e1ef8314de9b66b22630b7bd99e : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0fc6a83e0e760f5f5f0fdea3b38c8b43d9053c4c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-18 18:38:21.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, still needs ack from Renat.","accounts_in_message":[],"_revision_number":1},{"id":"c0f71d9f797d0be01d9d92f99296d4523bcdb7d7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-19 04:42:42.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f0bc2fba85ac5a051cf777632d32866f4e4f0847","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-20 14:08:19.000000000","message":"Patch Set 1: Code-Review-1\n\nAre we okay that this adds features and changes requirements as a Z bump? I would have a tendency to say no. Don\u0027t hesitate to overrule me if I am wrong there.","accounts_in_message":[],"_revision_number":1},{"id":"f77a114dc6dfaef6fe4755058afa458d0c1624c9","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-20 14:48:22.000000000","message":"Patch Set 1:\n\n\u003e Are we okay that this adds features and changes requirements as a Z\n \u003e bump? I would have a tendency to say no. Don\u0027t hesitate to overrule\n \u003e me if I am wrong there.\n\nhuh, that is a good point. I didn\u0027t notice there were so many backported but unreleased changed. That isn\u0027t great. I guess I\u0027ll bump this to a feature release.","accounts_in_message":[],"_revision_number":1},{"id":"df313e22886fa99482b6048876db2284c667e284","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-20 14:53:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d2f4076676bd0499ca634f4baa8bb6355eefa8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 15:15:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f505385649b74803a5e0a1e895b9b2b3 : SUCCESS in 8m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8252e67b307049d6b401619ce201dc6a : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/450c625bb824442a9bb03be3161a332d : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"969c6f772dfb96f9077d591093a6ca502bd50bcf","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-21 10:11:22.000000000","message":"Patch Set 2: Code-Review+1\n\nIt is definitely against the stable policy to backport features to stable branches, but then Mistral does not assert it follows the stable policy...\n\nSo I guess this is fine, although I wish Mistral would follow the common stable policy.\n\nWe\u0027ll need a confirmation from Renat to go ahead despite the warnings :)","accounts_in_message":[],"_revision_number":2},{"id":"07bde0b8b67b8524bd214fa6065424981b3e448e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-21 10:21:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bb5b670508a9620bfb9ce42492f59ea85e0da131","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-21 16:33:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1aa7a7cb390dd92e7d540e8502b74e85b3271fc3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-22 13:41:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c417630c6be0c1d9d3161adafecc86d5c6d72a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 13:41:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a73f0ddc8826b837ff3b6641027cdd8cec611606","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 13:55:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9d19062dfcff71d0b2d629b94e2d3c3473739215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 13:55:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a6b9d743dd4e44a3e41c4cfe233df1 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7ee613f70840ffb878cd1e034a3d9a : SUCCESS in 9m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef99b60c855841c18e7a397bb1553b3f : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ce63011bbefe8ee6aaa30f215682fe170ccad10","revisions":{"7e06ee9b1d9a1d5bf82721c9842b08f49cfbc414":{"kind":"REWORK","_number":1,"created":"2019-11-18 14:37:52.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/86/694786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/694786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/694786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/694786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/694786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/694786/1"}}},"commit":{"parents":[{"commit":"bca6c7b82ca29b040c8e61b9ce5201c1dd2de799","subject":"Merge \"Release networking-hyperv 7.0.1 (rocky)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bca6c7b82ca29b040c8e61b9ce5201c1dd2de799"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-18 14:37:32.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-18 14:37:32.000000000","tz":0},"subject":"Bugfix release for Mistral Stein","message":"Bugfix release for Mistral Stein\n\nChange-Id: I2c0ca25ff960d5d969746df6a851253d4fa69faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e06ee9b1d9a1d5bf82721c9842b08f49cfbc414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e06ee9b1d9a1d5bf82721c9842b08f49cfbc414"}]},"branch":"refs/heads/master"},"8ce63011bbefe8ee6aaa30f215682fe170ccad10":{"kind":"REWORK","_number":2,"created":"2019-11-20 14:53:55.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/86/694786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/694786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/694786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/694786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/694786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/694786/2"}}},"commit":{"parents":[{"commit":"e378cddcd030b734f8b7d4d57e7103f46264ba3c","subject":"Unpin keyring to unbork twine validate check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e378cddcd030b734f8b7d4d57e7103f46264ba3c"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-20 14:53:26.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-20 14:53:26.000000000","tz":0},"subject":"Feature release for Mistral Stein","message":"Feature release for Mistral Stein\n\nChange-Id: I2c0ca25ff960d5d969746df6a851253d4fa69faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ce63011bbefe8ee6aaa30f215682fe170ccad10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ce63011bbefe8ee6aaa30f215682fe170ccad10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
