)]}'
{"id":"openstack%2Freleases~694955","triplet_id":"openstack%2Freleases~master~I904ea3735daaa8892edc54be4a9611beb66fdfe2","project":"openstack/releases","branch":"master","topic":"unpin-keyring","hashtags":[],"change_id":"I904ea3735daaa8892edc54be4a9611beb66fdfe2","subject":"Unpin keyring to unbork twine validate check","status":"MERGED","created":"2019-11-19 09:15:38.000000000","updated":"2019-11-19 10:06:41.000000000","submitted":"2019-11-19 10:06:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694955-1574158001220-2b47506d","meta_rev_id":"518550122ba8cf5a8085fdf4aef84a434576ff2a","_number":694955,"virtual_id_number":694955,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 10:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-11-19 09:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-11-19 09:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-11-19 09:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 09:53:54.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-11-19 10:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cf909021d0fe134ce1e2b55fe85004dc1d9a72e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-19 09:15:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"700dbd0d37485f99ea895171ab9cdaf6f250d090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:37:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374573cbbf6643e0a71d4dd3c2ff954c : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7990d362d802435ab4ca3890af9fec11 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f214882500914eb29ad3e981ad1d6ced : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e0406c81cf4e1e8ed60a118cb739d3 : SUCCESS in 9m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/273002f4a26944ae8aaa6d2df92e041f : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/925264d5146e488fac7a3d4543a00990 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3c9a6df9d4603fac8400a08d5c1f9c7480f14d47","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-19 09:41:18.000000000","message":"Patch Set 1:\n\nLooking at before/after logs, newer keyring ends up installing a couple more deps, but nothing like the full GUI support that the pin was added to prevent.","accounts_in_message":[],"_revision_number":1},{"id":"a09177500b159d91e451812aa1bd1506d0a7b5f2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-19 09:46:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks okay. This is better than staying in the past anyway. Passes CI with newly updated version, let\u0027s see if there are hidden impact by merging soon ! :p","accounts_in_message":[],"_revision_number":1},{"id":"92b1671dfa95052c573362b76fe6344a89d830a8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-19 09:53:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98b9205354aef841d6676a711f0451ea159cbfe4","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-19 09:53:54.000000000","message":"Patch Set 1: Workflow+1\n\nThis is blocking people, and we won\u0027t have reviewers right now (KubeCon). Let\u0027s merge it now.","accounts_in_message":[],"_revision_number":1},{"id":"03ff67d370612dfb76101905d070c746ad9ebfed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:54:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e2ea6b6a631628cae1afa8ace1c8d094a3852c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:06:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6803bc978449a795e8de95b823f4f7 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69249bf4abfc421f931b869a6eac900d : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22e059459221479b888e0dd5de10d96a : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1fca95fae649f08c725d14295acb24 : SUCCESS in 10m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa4dc8bd4b634640b5ded3e96f2390f2 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"54050ff1ea954b25451b70d7c8ae542f553f5a28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:06:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e378cddcd030b734f8b7d4d57e7103f46264ba3c","revisions":{"e378cddcd030b734f8b7d4d57e7103f46264ba3c":{"kind":"REWORK","_number":1,"created":"2019-11-19 09:15:38.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/55/694955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/694955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/694955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/694955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/694955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/694955/1"}}},"commit":{"parents":[{"commit":"7202fe1edc69a82eab041ccaf86472638766f049","subject":"Merge \"Make Dmitriy release liaison for OSA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7202fe1edc69a82eab041ccaf86472638766f049"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-19 09:11:52.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-19 09:11:52.000000000","tz":60},"subject":"Unpin keyring to unbork twine validate check","message":"Unpin keyring to unbork twine validate check\n\nAs reported by Tim Burke, just-released twine 3.0.0 requires\nkeyring\u003e\u003d15.1, but we have a 3-year-old keyring\u003d\u003d7.3 pin in\nreleases requirements.\n\nThe pin was put to avoid installing a full desktop (through a\nDBUS dependency introduced in \u003e7.3) but this seem to have been\nfixed between now and then.\n\nChange-Id: I904ea3735daaa8892edc54be4a9611beb66fdfe2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e378cddcd030b734f8b7d4d57e7103f46264ba3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e378cddcd030b734f8b7d4d57e7103f46264ba3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
