)]}'
{"id":"openstack%2Freleases~695162","triplet_id":"openstack%2Freleases~master~I2298ef11e15ce7ade4c232899df1c1d2225c9c5d","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I2298ef11e15ce7ade4c232899df1c1d2225c9c5d","subject":"Fix ssh host addr link.","status":"MERGED","created":"2019-11-20 10:17:01.000000000","updated":"2019-11-21 10:07:43.000000000","submitted":"2019-11-21 10:07:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695162-1574330863509-fe519b9b","meta_rev_id":"2fdbf20a33ec121be10bfaa39ec8c7b6edd5a7d3","_number":695162,"virtual_id_number":695162,"owner":{"_account_id":30717,"name":"wangfaxin","email":"wangfaxin@inspur.com","username":"wangfaxin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 10:07:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-11-21 09:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-11-20 14:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-11-21 09:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 14:20:29.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-11-21 09:54:41.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-21 10:07:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0143f4f294b26e3b2da267367e232ceec309a55","author":{"_account_id":30717,"name":"wangfaxin","email":"wangfaxin@inspur.com","username":"wangfaxin"},"date":"2019-11-20 10:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae2bee715e4a39bc48970cf3527777c40d61ab0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 10:30:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbab08696ce24877b8281d669bb39f6e : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb6803ed261465e867d74ea434be9e2 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e1bcbb59917418a8f653261bd658ea8 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df98454204148a2a847ef92388b9d11 : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37a930bf41854682a2b5325b6d49704c : SUCCESS in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2d24eb10b90141b8ab25bfdaad8bd479 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e86a575e85a3d94d7ba98c4dca02689da8c58c2f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-20 14:20:29.000000000","message":"Patch Set 1: Code-Review+2\n\ngood catch","accounts_in_message":[],"_revision_number":1},{"id":"2e8ba4f71ed2c4b175aa8a90b3b272dba5d37e9e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-21 09:54:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7443d652ec12ab1e931a3615a08e3b335cd42668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 09:54:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c699ea5e4432485e278704832fb183028bb947ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:07:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2fdbf20a33ec121be10bfaa39ec8c7b6edd5a7d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 10:07:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8402a60645430da4163bd05dd02795 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caca3a63d75a4ec898899b1c4ccdd77f : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d23ad857a5bb494d83a7718cb8435c89 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55421794fbf464a8fc76af8875e12a0 : SUCCESS in 9m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea091d977dbf48bbac3a00583212be48 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d3a0b03840e70ddb9ae6f7c7d870e7eef8b157f","revisions":{"8d3a0b03840e70ddb9ae6f7c7d870e7eef8b157f":{"kind":"REWORK","_number":1,"created":"2019-11-20 10:17:01.000000000","uploader":{"_account_id":30717,"name":"wangfaxin","email":"wangfaxin@inspur.com","username":"wangfaxin"},"ref":"refs/changes/62/695162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/695162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/695162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/695162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/695162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/695162/1"}}},"commit":{"parents":[{"commit":"e378cddcd030b734f8b7d4d57e7103f46264ba3c","subject":"Unpin keyring to unbork twine validate check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e378cddcd030b734f8b7d4d57e7103f46264ba3c"}]}],"author":{"name":"wangfaxin","email":"wangfaxin@inspur.com","date":"2019-11-20 10:15:03.000000000","tz":480},"committer":{"name":"wangfaxin","email":"wangfaxin@inspur.com","date":"2019-11-20 10:16:54.000000000","tz":480},"subject":"Fix ssh host addr link.","message":"Fix ssh host addr link.\n\nChange-Id: I2298ef11e15ce7ade4c232899df1c1d2225c9c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d3a0b03840e70ddb9ae6f7c7d870e7eef8b157f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d3a0b03840e70ddb9ae6f7c7d870e7eef8b157f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
