)]}'
{"id":"openstack%2Freleases~695653","triplet_id":"openstack%2Freleases~master~Iebe99745b39979b88cc98515e72fec878caa9805","project":"openstack/releases","branch":"master","topic":"dnm-test-fstrings","hashtags":[],"change_id":"Iebe99745b39979b88cc98515e72fec878caa9805","subject":"[DNM] Example introducing f-strings","status":"ABANDONED","created":"2019-11-22 10:36:28.000000000","updated":"2019-11-22 13:40:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d829d55a363b08c108e1840e25dba1d5e139b63","_number":695653,"virtual_id_number":695653,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-22 11:10:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 11:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07eecd1460fc1a0238a7f712d3312a5edfd88e60","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-22 10:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"244844362c7ffe05e8623bd130b1d351cf18883f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 10:49:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdcf850509b5447b95e7cb8192cfb06f : FAILURE in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c258c22e231446f98b77bc8c2b89cf43 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a12da8954c547fca3886ee4539e2cf6 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2c2d6396da4da79986a025fd1ce44c : SUCCESS in 11m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8141ec85f6744e67a1591ab488650bbd : SUCCESS in 4m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/84632deac81e4748abb9840ffcdf003d : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"59aed5516ccdd13c2b449ca668705ffeca50e60b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-22 10:56:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"141e5654ea61fc629766d29ebd9bbee0c061955b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 11:10:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107afb110b8540ec8b79e7e6b0a3badc : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0fcb04e6e241e699b9a65141282103 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4f182a0ff794d93a944f88c1367637f : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37a52c13adf41918ba0d0545823186d : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d0b4f53ef174ebfac06c742fb17370d : SUCCESS in 5m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9828a6c1dddb496e82a9b2521450e209 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b59d6498673c7cab054158972b0849e3c780976f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-22 13:40:15.000000000","message":"Abandoned\n\nThat was a practical example.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98b6c66e4aba44ff021f8b6832b6f23fc754d466","revisions":{"a9192387cc8188284cf060fae1ed7cf79fad0116":{"kind":"REWORK","_number":1,"created":"2019-11-22 10:36:28.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/53/695653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/695653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/695653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/695653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/695653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/695653/1"}}},"commit":{"parents":[{"commit":"781faaf09754d7167855b466ee7decdcfdc8f295","subject":"Merge \"Update process for R+1 week\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/781faaf09754d7167855b466ee7decdcfdc8f295"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-22 10:34:33.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-22 10:34:33.000000000","tz":60},"subject":"[DNM] Example introducing f-strings","message":"[DNM] Example introducing f-strings\n\nDo not merge! This is a demo change.\n\nTry using f-strings. It should fail in patchset 1, but pass if we\nexplicitly bump pyflakes in patchset 2.\n\nChange-Id: Iebe99745b39979b88cc98515e72fec878caa9805\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9192387cc8188284cf060fae1ed7cf79fad0116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9192387cc8188284cf060fae1ed7cf79fad0116"}]},"branch":"refs/heads/master"},"98b6c66e4aba44ff021f8b6832b6f23fc754d466":{"kind":"REWORK","_number":2,"created":"2019-11-22 10:56:37.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/53/695653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/695653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/695653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/695653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/695653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/695653/2"}}},"commit":{"parents":[{"commit":"781faaf09754d7167855b466ee7decdcfdc8f295","subject":"Merge \"Update process for R+1 week\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/781faaf09754d7167855b466ee7decdcfdc8f295"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-22 10:34:33.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-22 10:37:07.000000000","tz":60},"subject":"[DNM] Example introducing f-strings","message":"[DNM] Example introducing f-strings\n\nDo not merge! This is a demo change.\n\nTry using f-strings. It should fail in patchset 1, but pass if we\nexplicitly bump pyflakes in patchset 2.\n\nChange-Id: Iebe99745b39979b88cc98515e72fec878caa9805\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98b6c66e4aba44ff021f8b6832b6f23fc754d466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98b6c66e4aba44ff021f8b6832b6f23fc754d466"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
